)]}'
{"id":"openstack%2Fglance_store~777615","triplet_id":"openstack%2Fglance_store~stable%2Fstein~Id9bc9172560421e3ebcbb3842da367ffd9a6bfd4","project":"openstack/glance_store","branch":"stable/stein","topic":"native-zuulv3-migration","hashtags":[],"change_id":"Id9bc9172560421e3ebcbb3842da367ffd9a6bfd4","subject":"zuul: glance_store-src-ceph-tempest replaces a legacy job","status":"ABANDONED","created":"2021-02-25 16:09:47.000000000","updated":"2022-05-11 14:32:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2bd2c7262d6700dd6e1e03ff601f7a04b21a8830","_number":777615,"virtual_id_number":777615,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-09 14:15:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2021-07-09 16:48:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2021-08-11 05:24:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2021-08-11 05:24:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-25 17:54:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-09 16:48:39.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2021-08-11 05:24:35.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"4b6f39a82f8952bdc2d8929b392ee94d2470ff95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-25 16:09:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48b0833b264bb1dfdd97f2c337eefc9abcf61ca2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 17:54:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/99698df5060f424ea441ed10c60550c3 : SUCCESS in 1h 36m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d5c214ffeac34e4ebe7563a0754330ec : SUCCESS in 1h 29m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/691cc5318f994b90802c2f36624fde05 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1b95bc862fb46239d34a55fe67b5496 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b0b6407520a442cfa6234d08aec4edc9 : FAILURE in 3m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/692d327cc4c7443c8c0da73e554b9f8e : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecbc3800ee8d4cf9ae7b2efc84da4879 : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e8189172cf34889b3c90518ab9f7e32 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f0256cdff344bde84d3c71d7ab8ba8d : SUCCESS in 4m 19s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/9e3d849fd2dd4692826af186ada11502 : FAILURE in 1h 12m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2027b96c59b966e5680fa999d248f44d6b5ca4b9","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-06-09 10:22:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c1d9e4d7bddaacdb8276c2813978a10be51c8aa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 11:48:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/cf573820b54a4090a81730b7479d9321 : SUCCESS in 1h 24m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e6d5b06b407444c9549f9b2bfe7e702 : SUCCESS in 1h 01m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86110c6940684f8cabd7851f8ff809ad : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8ff58725bc047458b7e10a72cee91f7 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ddc41d05dbb64dfcb7c8ed69c1fbcc0e : FAILURE in 3m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dfa3713ad711422199e27d69edda4e10 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d8a446d55e642b099b8f08b070461cb : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7c809bfac914a85b7efd83ed6775cdf : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57bb18fe97d24e75b5fbed2e1bec3c32 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"0193feb99f94a440e10d6b21f58d98ac1a68f24a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-06-09 12:34:57.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"561ab34c19e40d07f551e84ef17c0c713e6abd0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 14:15:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/24b4e3010fc44b5dbeb189ed1e18b898 : SUCCESS in 1h 31m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/369db45da60d45f994cfe1897b51b410 : SUCCESS in 1h 35m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4da18804e5f400cbf8f08da6e2e1840 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd5d23c32a574fe4bd3ff58e26d2ce73 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/16dba5c74c494c15af5c1f1645cc0cb3 : SUCCESS in 4m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d49a4e2b7d2146a99629bade1a2dce74 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a673e18adc0b46c7b54033ccab02e69f : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e6f0d8a8b2ca4500977ad86a07f86828 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/151a75efefa7491a8b5a97b1f4056fd7 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":2},{"id":"113fee31a925883396be2a488f10fac324c7d48b","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-07-09 16:48:39.000000000","message":"Patch Set 2: Code-Review+2\n\nyeii, last one ready to go.","accounts_in_message":[],"_revision_number":2},{"id":"9768dc843ae6a645a97cae45551db0a68171be5a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-11 05:24:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"289502ef45318152755e8b410028d99d745080a0","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-08-15 06:46:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ac2527f4fecafd7fe158ec132c855273c960ca61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-15 08:36:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c9be903436f1463a859caa7f9dd5cd9c : SUCCESS in 1h 27m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/63a8b925d4e547e68332aea79f78f0d0 : SUCCESS in 1h 19m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c76a38b425c4323a75112eabddd2334 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a891064f784472ca562edc879eb7637 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cceef43b3566489f94b4ca7fff45807c : SUCCESS in 4m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8fcfd0eb938940c6a7e9235639823b7f : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abdc6e659e94410f938f22729e6ee389 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca73e753f1be491089fbc51ca838d37f : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73bdcc76c9ec47c28233ce9d695180eb : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"2bd2c7262d6700dd6e1e03ff601f7a04b21a8830","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-11 14:32:35.000000000","message":"Abandoned\n\nThis does not fit with our backport criteria [0]\n\n[0] https://docs.openstack.org/project-team-guide/stable-branches.html#appropriate-fixes","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"99eb10d3ea6fc54224ae325350ff2d1309384b1f","revisions":{"5381551487b02ca44b27ac0b47ec76b41c63754d":{"kind":"REWORK","_number":1,"created":"2021-02-25 16:09:47.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/15/777615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/15/777615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/15/777615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/15/777615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/15/777615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/15/777615/1"}}},"commit":{"parents":[{"commit":"27e421af26cd1bcc68746a0beadb541fb8d647be","subject":"Remove sheepdog tests from zuul config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/27e421af26cd1bcc68746a0beadb541fb8d647be"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-07-27 10:52:48.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2021-02-25 16:09:12.000000000","tz":60},"subject":"zuul: glance_store-src-ceph-tempest replaces a legacy job","message":"zuul: glance_store-src-ceph-tempest replaces a legacy job\n\nIt replaces legacy-tempest-dsvm-full-ceph-plugin-src-glance_store\nwhich is defined inside openstack-zuul-jobs.\nIt is really a variant of the devstack-plugin-ceph-tempest job.\n\nRemove .zuul.yaml from the ignored file list as in other jobs.\n\nConflicts:\n\t.zuul.yaml\nThe conflict originates from the other unrelated changes in the file.\nBranch: specific change: use devstack-plugin-ceph-tempest\ninstead of devstack-plugin-ceph-tempest-py3 as parent job.\n\nChange-Id: Id9bc9172560421e3ebcbb3842da367ffd9a6bfd4\n(cherry picked from commit 1f50f31c14803ea2eec021b3deb287b5a9facc27)\n(cherry picked from commit bbd97858d9c5b8d2ddc14accb28fec855254c926)\n(cherry picked from commit 79e043af2a86f4547f8e7bad6be90ff4542e005e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/5381551487b02ca44b27ac0b47ec76b41c63754d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/5381551487b02ca44b27ac0b47ec76b41c63754d"}]},"branch":"refs/heads/stable/stein"},"99eb10d3ea6fc54224ae325350ff2d1309384b1f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-06-09 12:34:57.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/15/777615/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/15/777615/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/15/777615/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/15/777615/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/15/777615/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/15/777615/2"}}},"commit":{"parents":[{"commit":"0503776e8f97e7b174d3dc4eef39d235b769436e","subject":"remove bandit from testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0503776e8f97e7b174d3dc4eef39d235b769436e"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-07-27 10:52:48.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2021-06-09 12:34:26.000000000","tz":120},"subject":"zuul: glance_store-src-ceph-tempest replaces a legacy job","message":"zuul: glance_store-src-ceph-tempest replaces a legacy job\n\nIt replaces legacy-tempest-dsvm-full-ceph-plugin-src-glance_store\nwhich is defined inside openstack-zuul-jobs.\nIt is really a variant of the devstack-plugin-ceph-tempest job.\n\nRemove .zuul.yaml from the ignored file list as in other jobs.\n\nConflicts:\n\t.zuul.yaml\nThe conflict originates from the other unrelated changes in the file.\nBranch: specific change: use devstack-plugin-ceph-tempest\ninstead of devstack-plugin-ceph-tempest-py3 as parent job.\n\nChange-Id: Id9bc9172560421e3ebcbb3842da367ffd9a6bfd4\n(cherry picked from commit 1f50f31c14803ea2eec021b3deb287b5a9facc27)\n(cherry picked from commit bbd97858d9c5b8d2ddc14accb28fec855254c926)\n(cherry picked from commit 79e043af2a86f4547f8e7bad6be90ff4542e005e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/99eb10d3ea6fc54224ae325350ff2d1309384b1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/99eb10d3ea6fc54224ae325350ff2d1309384b1f"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
