)]}'
{"id":"openstack%2Fglance_store~786420","triplet_id":"openstack%2Fglance_store~stable%2Fwallaby~I82979c47ec9ec1dd6cc864d3827ab219177251f8","project":"openstack/glance_store","branch":"stable/wallaby","hashtags":[],"change_id":"I82979c47ec9ec1dd6cc864d3827ab219177251f8","subject":"Pass multipath config while creating connector object","status":"MERGED","created":"2021-04-15 16:35:58.000000000","updated":"2021-04-19 08:38:28.000000000","submitted":"2021-04-19 08:37:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786420","cherry_pick_of_change":786271,"cherry_pick_of_patch_set":4,"meta_rev_id":"668d2fc2831e64ef7f2642beda5d13597d1c764e","_number":786420,"virtual_id_number":786420,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-19 08:37:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-04-19 07:10:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-04-19 07:10:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-15 19:01:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-19 07:10:02.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"0afcfd7dd11eba75613674ece8e9354d77196fb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-04-15 16:35:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"16696eee9903b13cc96ee6ead8f3c429740ea99f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 19:01:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b82550c4b47c49ddbe78e99cde9cbea9 : SUCCESS in 1h 49m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c63c48af933545b68d71bc64f720a541 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b1c936387b1493bb40568a27a9283b5 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ca28bbcc84e409db8dbc196a73e4006 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a0dbd8deeb143d3b83bc4a2f24878be : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f17b54fabdf48c1b6d9cab73e1a4e01 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/960faaf01e164d32811e52d458dc4310 : SUCCESS in 4m 51s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/e6067cb00dda4172b1fc78872f2c154d : SUCCESS in 10m 58s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/06f7868986f8458da0d84e915b008681 : TIMED_OUT in 2h 05m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e2093a011322ce9da7b0b1bce2f99b581886b88b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-04-19 07:10:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean and straightforward backport.\nNinja approving as short on cores","accounts_in_message":[],"_revision_number":1},{"id":"febd5cee24e92dcf43974f53e78bda98aeb7b181","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 07:10:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6bd18af133b4cea41f5188c04e4ae6a2f81234a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 08:37:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0cb6b1487d7f4ee68d26945b315f3a33 : SUCCESS in 1h 25m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63923ec5e1ca486bb4bdd60c0438294c : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4933bf2e26fb402b857bca4c03e56d45 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d145f0529584489be4baa0425384c3a : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7f7b13e60814a9c8f5cbacf6f1c7843 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fab456ecb25e4041bf187b7d1f5f3ca0 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b44a177faddca34ab06c0e12d9e6e3f10f809015","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 08:37:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"668d2fc2831e64ef7f2642beda5d13597d1c764e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 08:38:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d4312de1aee40589f3ceeca721d55d4 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9bb857097e8847bf8c9c722f7ff5990dccb391cf","revisions":{"9bb857097e8847bf8c9c722f7ff5990dccb391cf":{"kind":"REWORK","_number":1,"created":"2021-04-15 16:35:58.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/20/786420/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/20/786420/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/20/786420/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/20/786420/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/20/786420/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/20/786420/1"}}},"commit":{"parents":[{"commit":"ae9022cd3639bf3d0f482921d03b2b751f757399","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ae9022cd3639bf3d0f482921d03b2b751f757399"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-04-14 14:40:45.000000000","tz":-240},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-04-15 16:35:58.000000000","tz":0},"subject":"Pass multipath config while creating connector object","message":"Pass multipath config while creating connector object\n\nCurrently we only pass the configured multipath option when getting\nthe connector properties and do not pass it when creating the\nconnector object which in turn does the multipathing in os-brick.\nSo we are only passing multipathing option to cinder for exporting \u0026\nmapping the volume to paths but not while creating the connector object\nand hence doing multipathing partially.\nThis patch passes that option and fixes the stated issue.\n\nRelated-Bug: #1863983\n\nCo-Authored-By: Gorka Eguileor \u003cgeguileo@redhat.com\u003e\n\nChange-Id: I82979c47ec9ec1dd6cc864d3827ab219177251f8\n(cherry picked from commit 04e5ead7c000211a4c10104ed2bb65c9df7681ae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/9bb857097e8847bf8c9c722f7ff5990dccb391cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/9bb857097e8847bf8c9c722f7ff5990dccb391cf"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
