)]}'
{"id":"openstack%2Fglance_store~790082","triplet_id":"openstack%2Fglance_store~stable%2Fwallaby~I064695dd1a5612ef750b4280c7d34534fab530bf","project":"openstack/glance_store","branch":"stable/wallaby","hashtags":[],"change_id":"I064695dd1a5612ef750b4280c7d34534fab530bf","subject":"Drop lower-constraints job","status":"ABANDONED","created":"2021-05-06 14:58:46.000000000","updated":"2021-08-12 09:20:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"76b5278600f68b9c62a0c22819a77afc48284749","_number":790082,"virtual_id_number":790082,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":0,"date":"2021-08-12 09:06:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2021-08-12 09:06:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-05-07 19:12:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-11 21:53:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2021-08-12 09:07:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-06 17:14:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-07 13:43:38.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-05-11 21:53:47.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-08-12 09:06:20.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"562cec9f671e97949463a25d0d5816f96ebf40cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-05-06 14:58:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1672a9d553634281c2ba8ed819356adbe509d2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 17:14:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f46de593d3cd409a956580861819829c : FAILURE in 1h 23m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0b2efc75605465296e5ac1d7e1c6160 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2ec6d18395d4191a356b12517a9e67a : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42d9ead5b88c4c2684371f2b56bbbd49 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/092af2323ecb4acb8397d85309c30471 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53d27b99af724235b26432fb960b3fb8 : SUCCESS in 5m 22s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/86402a359f174c2aba06c6481807b3dd : SUCCESS in 9m 18s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/cce09eeaf1604d87a176ce083ffc8184 : TIMED_OUT in 2h 09m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67632fd72aec1478d5a944beda4d1c2ee70cc44d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-05-06 17:25:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2851dcc950ace89a9f3f4ae77b593002592d832d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 19:35:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/59c08bace24d4fa980e70f958724192d : SUCCESS in 1h 46m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad1ed4a3de25436abf5947d9a7017d4e : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a589d2a13bd44a1b3b3b48f5b4a867d : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9914bc602c344f0897d69bccfe89efbd : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2a1600ffb2148d494bfbb33d0d58ecb : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afe15fa8422643eaa2e93b11c5908bf8 : SUCCESS in 3m 57s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/d4433c9d6012404587768d81cd57ffb4 : SUCCESS in 8m 14s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/b117323c11b94d49ae7e385a4434fb1a : TIMED_OUT in 2h 07m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"719a97529c7b635707cddef3ee763e071da3b02b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-05-07 13:43:38.000000000","message":"Patch Set 1: Review-Priority+2\n\nI still think we need to do this on all our master and stable trees. This is a good start, but I think we need to do this on master for glance_store as well as stables. If we\u0027re not going to do that, then we need to fix this instead of just removing it.","accounts_in_message":[],"_revision_number":1},{"id":"3fd9f2b4949597a3632550a82937491c758ca077","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-05-07 13:49:19.000000000","message":"Patch Set 1: Code-Review-1 -Review-Priority\n\nOh, I just noticed, you need to remove the lower-constraints.txt and the tox.ini section as well","accounts_in_message":[],"_revision_number":1},{"id":"65019af0bff9b491d5cf3c2d3fd192c938613a8d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-05-07 14:00:34.000000000","message":"Patch Set 1:\n\n\u003e Oh, I just noticed, you need to remove the lower-constraints.txt and the tox.ini section as well\n\nThis is what we merged in glance and backported back further:\n\nhttps://review.opendev.org/c/openstack/glance/+/782769","accounts_in_message":[],"_revision_number":1},{"id":"adc267d3ce482f0aa992a685718d89e5b6784243","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-05-07 19:05:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fac3420736098a4ff51b33f5792659b9a025d79d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-05-07 19:12:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"478e1c596059f9afe8a533d80fd4a02a9b5f25b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 20:54:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9939bc07b5b5427ba3824c4859a51a1e : SUCCESS in 2m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e5e13fe56c6458e9c1658eed855f281 : FAILURE in 1h 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8094d3372484c4d95826a008f847be5 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6958c299045b4a139e4d2097000af09d : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5b01c429d5e49e2b2efde20dd549b47 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f1eda6134834ecc9f5452c6ac84622c : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23795f101fb64b1aa7f19ba93d6ee1f7 : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c177bc6c2a047328ec4ca736335d3a9 : SUCCESS in 4m 29s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/6b8d65efa94a4b9bab46dc6a3a1eb8c8 : FAILURE in 16m 31s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/cd8152d043a448758596f27fd26caff6 : SUCCESS in 1h 47m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"71da6fb02e2b070fa9d8d21084cc80187a66301f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-05-11 21:53:47.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e64f93d83e81cafb18a04fc42a74fc5a60faddc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 23:49:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/04ba43a941c64d34955cf5eb85d0df2b : SUCCESS in 2m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0012ae3856ae476e9f930eaa2562ee26 : SUCCESS in 1h 42m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93e15b25399c4a2e84dae3a780e9d383 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fd9e9c14b98416d8bc93cadcdbd7e38 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b473b9be51b84346886969621e71ee7a : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05af5ec2543c46be8b417b710775ef88 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71624e9cadad47be89c2496a78a7592f : SUCCESS in 3m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eefabe786b9d46fbb9f201d4757a6004 : SUCCESS in 4m 00s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/9fc2478b6dd8458e85a4f0bfc6db0249 : SUCCESS in 8m 22s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/837bca3e39414372b7dd5f225bdb4638 : SUCCESS in 1h 50m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"36e25668356af8eef6ae99f593e93260d011e667","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-12 09:06:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"91e6e8e86b9a172116046aff5120de118214835e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 09:06:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b309b2d0ce024bf0205389742c0da1fd2b20ece4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-12 09:07:05.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"2b5b88e32d79d95d35cf7bfe64f8a9d1ec69f231","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-12 09:08:03.000000000","message":"Abandoned\n\nProposing this to master and the we will backport it","accounts_in_message":[],"_revision_number":2},{"id":"76b5278600f68b9c62a0c22819a77afc48284749","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-12 09:20:06.000000000","message":"Patch Set 2:\n\nAbandoned in favor of https://review.opendev.org/c/openstack/glance_store/+/804369","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2d91da2670fe6dc47d0f538e43baa3d6404a9e1d","revisions":{"40534e1ed8421afd96a5adbda724344cfeefc0d6":{"kind":"REWORK","_number":1,"created":"2021-05-06 14:58:46.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/82/790082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/82/790082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/790082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/790082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/790082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/82/790082/1"}}},"commit":{"parents":[{"commit":"9bb857097e8847bf8c9c722f7ff5990dccb391cf","subject":"Pass multipath config while creating connector object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/9bb857097e8847bf8c9c722f7ff5990dccb391cf"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-05-06 14:55:48.000000000","tz":-240},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-05-06 14:58:42.000000000","tz":-240},"subject":"Remove openstack-lower-constraints-jobs","message":"Remove openstack-lower-constraints-jobs\n\nChange-Id: I064695dd1a5612ef750b4280c7d34534fab530bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/40534e1ed8421afd96a5adbda724344cfeefc0d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/40534e1ed8421afd96a5adbda724344cfeefc0d6"}]},"branch":"refs/heads/stable/wallaby"},"2d91da2670fe6dc47d0f538e43baa3d6404a9e1d":{"kind":"REWORK","_number":2,"created":"2021-05-07 19:05:08.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/82/790082/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/82/790082/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/790082/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/790082/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/790082/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/82/790082/2"}}},"commit":{"parents":[{"commit":"9bb857097e8847bf8c9c722f7ff5990dccb391cf","subject":"Pass multipath config while creating connector object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/9bb857097e8847bf8c9c722f7ff5990dccb391cf"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-05-06 14:55:48.000000000","tz":-240},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-05-07 19:01:42.000000000","tz":-240},"subject":"Drop lower-constraints job","message":"Drop lower-constraints job\n\nChange-Id: I064695dd1a5612ef750b4280c7d34534fab530bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2d91da2670fe6dc47d0f538e43baa3d6404a9e1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2d91da2670fe6dc47d0f538e43baa3d6404a9e1d"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
