)]}'
{"id":"openstack%2Fglance_store~792747","triplet_id":"openstack%2Fglance_store~master~Ib281337e350ea38a2911721a3a5bba2148e7711b","project":"openstack/glance_store","branch":"master","topic":"cinder_volume_retry_status","hashtags":[],"change_id":"Ib281337e350ea38a2911721a3a5bba2148e7711b","subject":"Allow to check the expected volume status multiple times before throwing an exception.","status":"NEW","created":"2021-05-23 18:00:26.000000000","updated":"2021-07-20 07:45:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":7,"unresolved_comment_count":7,"has_review_started":true,"meta_rev_id":"4975a4bfb3eaae6e2fc75f563b86f21cb6cfb6dc","_number":792747,"virtual_id_number":792747,"owner":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-15 08:37:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":-1,"date":"2021-07-16 11:42:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-23 20:49:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-01 15:25:16.000000000","updated_by":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-07-16 11:42:18.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2021-07-20 07:45:40.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REMOVED"}],"messages":[{"id":"c4eccd681149b0c923b8d27d4ea1e7595d1ef0ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"date":"2021-05-23 18:00:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76a03b2db43798bb40f3402240cbc4e1b9cb53be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"date":"2021-05-23 18:06:20.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"70f2db66f716dee855cf9763fe051c5a21837073","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"date":"2021-05-23 19:11:01.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"baf1015ce3dbd879bec45f86cfd6e0061a7cd2cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"date":"2021-05-23 19:12:28.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"d5b72e6dde31c20d2651f77b0397dadf36ba84a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-23 20:49:23.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/42d70b6dbe0b444f8e967727335c91b2 : SUCCESS in 1h 15m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7e098e26f964d60a5a5e0c790d4bb05 : FAILURE in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/574a5e5a361840a2970e1d17f92f07c8 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81dbe0e864724bb6a8febc00bb2bb3b6 : FAILURE in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/446831f2e38b4af5b504ff7b178d4d8c : FAILURE in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c26feba18f54a9a934daf56e56cc75f : FAILURE in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76be626d59a24cd18d4e371f56ca0f2e : SUCCESS in 3m 57s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/c7f4c89e5a65426c8b8e288138781869 : FAILURE in 8m 50s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/5a115e3ee0484b5094b62ec8efa590ff : SUCCESS in 1h 30m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"681888c9eaad3b479a16b5fc4437e7d6d4ed8c03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"date":"2021-05-25 16:44:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"32d8b8faf4799e7756aac83f9c095b560671c5f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 18:45:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/77914491c1eb4fdabe6c43272051a851 : SUCCESS in 1h 17m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6de0ebc986bf4da7a2557ff65e0119e8 : FAILURE in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffc30bcc903844cd85fb8f4b62e0adf3 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/725b523008dd41e8a454d37610958ec4 : FAILURE in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07a76ac060c449f6b4985f41cad8c4d3 : FAILURE in 11m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75c3ee2251b3476c91b70488bd0403f7 : FAILURE in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dc69468fc594e0d87975c7d0ee01f07 : SUCCESS in 5m 26s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/c571b8f27e6741109012bd24103ba1ea : FAILURE in 18m 52s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/4c93b3deb268493c9e09297827477104 : SUCCESS in 1h 59m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c582d5233a323ab1ed25eac043b4cea93329c14b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"date":"2021-05-25 20:42:10.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"52281a27818b77c7550e95009c84cac6d1449d94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 22:52:07.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b01ca0be4594b06a59f05337344a35a : SUCCESS in 1h 39m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7f4cc633b5846e0b75f304065a179ee : FAILURE in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0df6fcd94ab4edebad153350d3e71bc : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94211a3526544a01ad98e2f2eae6f207 : FAILURE in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4a7395e5c01494db980c8a862dfd61a : FAILURE in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d00d67e18eb46f0a369275b0c1d4f77 : FAILURE in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27a901cd40c4415693bde773e113bd27 : SUCCESS in 3m 55s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/b6bf55bba26c4960b60778649c61dc63 : SUCCESS in 10m 36s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/13c17e01f89047ec9afb0bc73e847c06 : SUCCESS in 2h 05m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e90d79305927fde50fc6fb9e0680e9010f081d76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"date":"2021-05-26 14:29:01.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f7879aec266340c80543ecb2fa0add71a83a857e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 16:30:11.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/128559d471aa40ccaca8c68956367590 : SUCCESS in 1h 46m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/868832d813bc48e69c8a3875f6e4da56 : FAILURE in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a32087a7b61f4956a4d657bb8d5bf8ca : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9d008d235ec4e4288dd80fc936afc6f : FAILURE in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce0ab7e50b9b40358a05e0e244d584e3 : FAILURE in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/399c8ae067234575ad0082297df5e236 : FAILURE in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e49986f0f024b5c8eb5f285c2a09175 : SUCCESS in 4m 13s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/b32af07a1f2a4ee7a51246b01e160443 : FAILURE in 8m 49s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/6b1cc7b064004fd48e27b17d9d2c74c2 : SUCCESS in 1h 56m 44s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"42f7553f28fd58c0cabc327e704e780c229da042","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"date":"2021-05-26 16:45:34.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d4df0e4ec89488c4eb76a699e5bca05812a14ba0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"date":"2021-05-26 16:48:12.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"da1191cfaa43aa170d65db1820ab564115567228","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 18:37:47.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/67f9ea8fdaf34f269048a9c0b6bb74be : SUCCESS in 1h 38m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bf60af1a1c8405687cf129d61970ac7 : TIMED_OUT in 41m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f84c5340c185439796b483eb7d11ccad : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1976b1381e7e44ca89fca3109d7a16ad : TIMED_OUT in 41m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfe6b7382b504408807e93e4db10b255 : TIMED_OUT in 41m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1eb9ea8bace24797841edcbf2b0349e5 : TIMED_OUT in 40m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bea77cad83ec46099402c5d0d420f421 : SUCCESS in 5m 23s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/8c99a3f8437441c4ba5c3ceb109d8d76 : FAILURE in 8m 31s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/13b6297acec84021a2b6b3031deab1e7 : SUCCESS in 1h 43m 49s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"fd2bf8d1d96c1a5e4edde92fd02a25b8c5cd059b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"date":"2021-05-26 18:44:58.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"929fb6734e7f1cad4f2c49c39d6ac300c3c921ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 20:52:35.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/03a29e593c91450c8b3cf6a63398963e : SUCCESS in 1h 42m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec4aa78450a543769ca7f0d20ca52e45 : TIMED_OUT in 41m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3122bd4e3ccd4c6a99406c6a6d929c9a : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06363779436845598597ac99736c59fd : TIMED_OUT in 41m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f03aafb8b3494ec998dfe72f914a537c : TIMED_OUT in 41m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0221e364a3c4648ae77434c190d2899 : TIMED_OUT in 41m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8150aa3dcbc74b59a166f99b795f763d : SUCCESS in 3m 37s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/9b247b9f429e4ed8bb96463737a0b6ef : SUCCESS in 8m 34s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/8f1e40f3317a4d33a68ffe5b224b1e5b : TIMED_OUT in 2h 05m 26s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"aead4993ebc6ad0f326c628a412c19a9ad596e6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"date":"2021-05-27 09:35:50.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"5ea90e2d7e2c58ab2c8b8176d4ba470fd7141d6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"date":"2021-05-27 09:38:24.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"249b7b5e309957cbaa5ade2b0da4fb9a4d5578e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 11:49:38.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3867f0a45c6b42b28f15bb2cec11c7cb : SUCCESS in 1h 34m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/489dae50cacf4ce88a42485a24d4c8d6 : FAILURE in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecb53a88715f4ca7bc58f4d5a99b4fa1 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/679dfc7c2ee04433b02ab4724fac3f5d : FAILURE in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68534ac6f9bf48f095a042aa52208a95 : FAILURE in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d80e430c3f743fda7d120c353eb6625 : FAILURE in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d3194e00d55464381fe4147d0abec4a : SUCCESS in 3m 59s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/de5219e2d077466ea0a1c11ccd1855e3 : FAILURE in 7m 55s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/49566dc3a733477892cf79888976ad42 : TIMED_OUT in 2h 05m 43s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"0e1764f15b1475e5b23850510e416f08c7cce50b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"date":"2021-05-27 11:56:51.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"c4c45acf20cadb07d909b1d539153a31600ca1f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 14:03:28.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74843341e0304703a8b52d8a2b1df4a7 : FAILURE in 1h 48m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2413fc30b6334df085b639793dd511e8 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a5589309aca4ec2966758f22a1fbdf9 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d73cce02dc524422a41e706257fba6ae : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea6f4b0640564375ba4422bc9c2dfa43 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52985664366f4d9d87b48da1e21c2949 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b089a66b1c764c908d0a9d7208645aff : SUCCESS in 3m 24s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/d5d2bd4b556c4022a084c24c3e1b0bd8 : FAILURE in 8m 25s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/d7181ad8394b40548875cc18f3b3dada : TIMED_OUT in 2h 05m 38s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b2e5420601172573c7e862b3c33c28f852c30512","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"date":"2021-06-15 06:50:35.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"ed3a6b936f97707f2cb1a202b7a8b9564e437cbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 08:37:12.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2cb5dc035c9d4f8db4c961800e088387 : SUCCESS in 1h 20m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a5c461cb26948b98c93dc31934fd889 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8efd6c8b888f478197b6b8b7748c88d3 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1164a5e8a8a5474aa273d01bd145ebf9 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f32a622c6574b07a521ec634675c358 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce36c2b1ec6b4b18b7ed6260c382f72a : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e27f307874a4188a0a8578c788708a8 : SUCCESS in 4m 29s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/1341218cb3ff4e5eb8d2d15b73cce850 : SUCCESS in 8m 59s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/2eb799f4be264a659a4bc5967067dfdb : SUCCESS in 1h 42m 09s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"adafd27178c2be81b4acda931ec28d2130144a90","author":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"date":"2021-06-20 11:07:10.000000000","message":"Patch Set 14:\n\nWhat\u0027s next?","accounts_in_message":[],"_revision_number":14},{"id":"a10a2e7477d25c61bf06d454556317666cbebbe4","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-07-16 11:42:18.000000000","message":"Patch Set 14: Code-Review-1\n\n(7 comments)\n\nThanks,\n\nI\u0027m pretty sure we have bug open for this behaviour, if not we really should. Please include into the commit message.\n\nFew pointers inline.","accounts_in_message":[],"_revision_number":14},{"id":"4975a4bfb3eaae6e2fc75f563b86f21cb6cfb6dc","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-07-20 07:45:40.000000000","message":"Removed reviewer Oleg Bondarev.","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"94c970faec576ce235cf4963d7987a9a145e7b04","revisions":{"518980e9a66ccb462a4732f587108ad44a5860cb":{"kind":"REWORK","_number":1,"created":"2021-05-23 18:00:26.000000000","uploader":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"ref":"refs/changes/47/792747/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/47/792747/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/47/792747/1"}}},"commit":{"parents":[{"commit":"04e5ead7c000211a4c10104ed2bb65c9df7681ae","subject":"Pass multipath config while creating connector object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/04e5ead7c000211a4c10104ed2bb65c9df7681ae"}]}],"author":{"name":"stress-t","email":"stress.t@gmail.com","date":"2021-05-23 15:11:15.000000000","tz":180},"committer":{"name":"stress_t","email":"stress.t@gmail.com","date":"2021-05-23 17:59:46.000000000","tz":180},"subject":"These changes allow you to check the expected volume status multiple times before throwing an exception.","message":"These changes allow you to check the expected volume status multiple times before throwing an exception.\n\nA case:\n- glance-api runs two or more times on separate servers with multiple threads.\n- using glance api in HA mode  and cinder as the backend.\n- we cannot use multipolar volumes.\n- Create a new virtual machine with temporary root disks. nova asks the glance api for the image, the glance api asks the cinder for the volume and the glance api try to mount the volume as a local disk to download nova. Cinder sends a response, and there we indicate a race condition for the two or more glance api servers. All answers from cinder include the volume status \"available\", but for the first glance of the api it is indeed \"available\" and for others it is \"in-use\".\n\nChange-Id: Ib281337e350ea38a2911721a3a5bba2148e7711b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/518980e9a66ccb462a4732f587108ad44a5860cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/518980e9a66ccb462a4732f587108ad44a5860cb"}]},"branch":"refs/heads/master"},"58801bc7c833b1206a333da9b9cf50041f096a49":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-05-23 18:06:20.000000000","uploader":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"ref":"refs/changes/47/792747/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/47/792747/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/47/792747/2"}}},"commit":{"parents":[{"commit":"04e5ead7c000211a4c10104ed2bb65c9df7681ae","subject":"Pass multipath config while creating connector object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/04e5ead7c000211a4c10104ed2bb65c9df7681ae"}]}],"author":{"name":"stress-t","email":"stress.t@gmail.com","date":"2021-05-23 15:11:15.000000000","tz":180},"committer":{"name":"stress_t","email":"stress.t@gmail.com","date":"2021-05-23 18:06:04.000000000","tz":180},"subject":"Allow to check the expected volume status multiple times before throwing an exception.","message":"Allow to check the expected volume status multiple times before throwing an exception.\n\nA case:\n- glance-api runs two or more times on separate servers with multiple threads.\n- using glance api in HA mode  and cinder as the backend.\n- we cannot use multipolar volumes.\n- Create a new virtual machine with temporary root disks. nova asks the glance api for the image, the glance api asks the cinder for the volume and the glance api try to mount the volume as a local disk to download nova. Cinder sends a response, and there we indicate a race condition for the two or more glance api servers. All answers from cinder include the volume status \"available\", but for the first glance of the api it is indeed \"available\" and for others it is \"in-use\".\n\nChange-Id: Ib281337e350ea38a2911721a3a5bba2148e7711b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/58801bc7c833b1206a333da9b9cf50041f096a49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/58801bc7c833b1206a333da9b9cf50041f096a49"}]},"branch":"refs/heads/master"},"ef12c1f281cf2c532ede27313a5e555218c5b71f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-05-23 19:11:01.000000000","uploader":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"ref":"refs/changes/47/792747/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/47/792747/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/47/792747/3"}}},"commit":{"parents":[{"commit":"04e5ead7c000211a4c10104ed2bb65c9df7681ae","subject":"Pass multipath config while creating connector object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/04e5ead7c000211a4c10104ed2bb65c9df7681ae"}]}],"author":{"name":"stress-t","email":"stress.t@gmail.com","date":"2021-05-23 15:11:15.000000000","tz":180},"committer":{"name":"stress_t","email":"stress.t@gmail.com","date":"2021-05-23 19:09:57.000000000","tz":180},"subject":"Allow to check the expected volume status multiple times before throwing an exception.","message":"Allow to check the expected volume status multiple times before throwing an exception.\n\nA case:\n- glance-api runs two or more times on separate servers with multiple threads.\n- using glance api in HA mode  and cinder as the backend.\n- we cannot use multiattach volumes.\n- Create a new virtual machine with temporary root disks. nova asks the glance api for the image, the glance api asks the cinder for the volume and the glance api try to mount the volume as a local disk to download nova. Cinder sends a response, and there we indicate a race condition for the two or more glance api servers. All answers from cinder include the volume status \"available\", but for the first glance of the api it is indeed \"available\" and for others it is \"in-use\".\n\nChange-Id: Ib281337e350ea38a2911721a3a5bba2148e7711b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ef12c1f281cf2c532ede27313a5e555218c5b71f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ef12c1f281cf2c532ede27313a5e555218c5b71f"}]},"branch":"refs/heads/master"},"61487839f610cda89e30bd5e5afb2cc5acd83785":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-05-23 19:12:28.000000000","uploader":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"ref":"refs/changes/47/792747/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/47/792747/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/47/792747/4"}}},"commit":{"parents":[{"commit":"04e5ead7c000211a4c10104ed2bb65c9df7681ae","subject":"Pass multipath config while creating connector object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/04e5ead7c000211a4c10104ed2bb65c9df7681ae"}]}],"author":{"name":"stress-t","email":"stress.t@gmail.com","date":"2021-05-23 15:11:15.000000000","tz":180},"committer":{"name":"stress_t","email":"stress.t@gmail.com","date":"2021-05-23 19:12:12.000000000","tz":180},"subject":"Allow to check the expected volume status multiple times before throwing an exception.","message":"Allow to check the expected volume status multiple times before throwing an exception.\n\nA case:\n- glance-api has been run two or more times on different servers with multiple threads.\n- using glance api in HA mode  and cinder as the backend.\n- we cannot use multiattach volumes.\n- Create a new virtual machine with temporary root disks. nova asks the glance api for the image, the glance api asks the cinder for the volume and the glance api try to mount the volume as a local disk to download nova. Cinder sends a response, and there we indicate a race condition for the two or more glance api servers. All answers from cinder include the volume status \"available\", but for the first glance of the api it is indeed \"available\" and for others it is \"in-use\".\n\nChange-Id: Ib281337e350ea38a2911721a3a5bba2148e7711b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/61487839f610cda89e30bd5e5afb2cc5acd83785"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/61487839f610cda89e30bd5e5afb2cc5acd83785"}]},"branch":"refs/heads/master"},"5969174b4aaa0f10c2dbc0e6670df5d80eff3f36":{"kind":"REWORK","_number":5,"created":"2021-05-25 16:44:31.000000000","uploader":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"ref":"refs/changes/47/792747/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/47/792747/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/47/792747/5"}}},"commit":{"parents":[{"commit":"04e5ead7c000211a4c10104ed2bb65c9df7681ae","subject":"Pass multipath config while creating connector object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/04e5ead7c000211a4c10104ed2bb65c9df7681ae"}]}],"author":{"name":"stress-t","email":"stress.t@gmail.com","date":"2021-05-23 15:11:15.000000000","tz":180},"committer":{"name":"stress_t","email":"stress.t@gmail.com","date":"2021-05-25 16:43:58.000000000","tz":180},"subject":"Allow to check the expected volume status multiple times before throwing an exception.","message":"Allow to check the expected volume status multiple times before throwing an exception.\n\nA case:\n- glance-api has been run two or more times on different servers with multiple threads.\n- using glance api in HA mode  and cinder as the backend.\n- we cannot use multiattach volumes.\n- Create a new virtual machine with temporary root disks. nova asks the glance api for the image, the glance api asks the cinder for the volume and the glance api try to mount the volume as a local disk to download nova. Cinder sends a response, and there we indicate a race condition for the two or more glance api servers. All answers from cinder include the volume status \"available\", but for the first glance of the api it is indeed \"available\" and for others it is \"in-use\".\n\nChange-Id: Ib281337e350ea38a2911721a3a5bba2148e7711b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/5969174b4aaa0f10c2dbc0e6670df5d80eff3f36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/5969174b4aaa0f10c2dbc0e6670df5d80eff3f36"}]},"branch":"refs/heads/master"},"866f4564a489b26119cec783ef8b972b442a541f":{"kind":"REWORK","_number":6,"created":"2021-05-25 20:42:10.000000000","uploader":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"ref":"refs/changes/47/792747/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/47/792747/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/47/792747/6"}}},"commit":{"parents":[{"commit":"04e5ead7c000211a4c10104ed2bb65c9df7681ae","subject":"Pass multipath config while creating connector object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/04e5ead7c000211a4c10104ed2bb65c9df7681ae"}]}],"author":{"name":"stress-t","email":"stress.t@gmail.com","date":"2021-05-23 15:11:15.000000000","tz":180},"committer":{"name":"stress_t","email":"stress.t@gmail.com","date":"2021-05-25 20:41:52.000000000","tz":180},"subject":"Allow to check the expected volume status multiple times before throwing an exception.","message":"Allow to check the expected volume status multiple times before throwing an exception.\n\nA case:\n- glance-api has been run two or more times on different servers with multiple threads.\n- using glance api in HA mode  and cinder as the backend.\n- we cannot use multiattach volumes.\n- Create a new virtual machine with temporary root disks. nova asks the glance api for the image, the glance api asks the cinder for the volume and the glance api try to mount the volume as a local disk to download nova. Cinder sends a response, and there we indicate a race condition for the two or more glance api servers. All answers from cinder include the volume status \"available\", but for the first glance of the api it is indeed \"available\" and for others it is \"in-use\".\n\nChange-Id: Ib281337e350ea38a2911721a3a5bba2148e7711b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/866f4564a489b26119cec783ef8b972b442a541f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/866f4564a489b26119cec783ef8b972b442a541f"}]},"branch":"refs/heads/master"},"8867a9dc04f4cd1898eb1197ad4c67eca2b54fc2":{"kind":"REWORK","_number":7,"created":"2021-05-26 14:29:01.000000000","uploader":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"ref":"refs/changes/47/792747/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/47/792747/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/47/792747/7"}}},"commit":{"parents":[{"commit":"04e5ead7c000211a4c10104ed2bb65c9df7681ae","subject":"Pass multipath config while creating connector object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/04e5ead7c000211a4c10104ed2bb65c9df7681ae"}]}],"author":{"name":"stress-t","email":"stress.t@gmail.com","date":"2021-05-23 15:11:15.000000000","tz":180},"committer":{"name":"stress_t","email":"stress.t@gmail.com","date":"2021-05-26 14:28:42.000000000","tz":180},"subject":"Allow to check the expected volume status multiple times before throwing an exception.","message":"Allow to check the expected volume status multiple times before throwing an exception.\n\nA case:\n- glance-api has been run two or more times on different servers with multiple threads.\n- using glance api in HA mode  and cinder as the backend.\n- we cannot use multiattach volumes.\n- Create a new virtual machine with temporary root disks. nova asks the glance api for the image, the glance api asks the cinder for the volume and the glance api try to mount the volume as a local disk to download nova. Cinder sends a response, and there we indicate a race condition for the two or more glance api servers. All answers from cinder include the volume status \"available\", but for the first glance of the api it is indeed \"available\" and for others it is \"in-use\".\n\nChange-Id: Ib281337e350ea38a2911721a3a5bba2148e7711b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/8867a9dc04f4cd1898eb1197ad4c67eca2b54fc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/8867a9dc04f4cd1898eb1197ad4c67eca2b54fc2"}]},"branch":"refs/heads/master"},"221320b71cfb0a5c9dfc06a175992b10fd572cdd":{"kind":"REWORK","_number":8,"created":"2021-05-26 16:45:34.000000000","uploader":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"ref":"refs/changes/47/792747/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/47/792747/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/47/792747/8"}}},"commit":{"parents":[{"commit":"04e5ead7c000211a4c10104ed2bb65c9df7681ae","subject":"Pass multipath config while creating connector object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/04e5ead7c000211a4c10104ed2bb65c9df7681ae"}]}],"author":{"name":"stress-t","email":"stress.t@gmail.com","date":"2021-05-23 15:11:15.000000000","tz":180},"committer":{"name":"stress_t","email":"stress.t@gmail.com","date":"2021-05-26 16:45:09.000000000","tz":180},"subject":"Allow to check the expected volume status multiple times before throwing an exception.","message":"Allow to check the expected volume status multiple times before throwing an exception.\n\nA case:\n- glance-api has been run two or more times on different servers with multiple threads.\n- using glance api in HA mode  and cinder as the backend.\n- we cannot use multiattach volumes.\n- Create a new virtual machine with temporary root disks. nova asks the glance api for the image, the glance api asks the cinder for the volume and the glance api try to mount the volume as a local disk to download nova. Cinder sends a response, and there we indicate a race condition for the two or more glance api servers. All answers from cinder include the volume status \"available\", but for the first glance of the api it is indeed \"available\" and for others it is \"in-use\".\n\nChange-Id: Ib281337e350ea38a2911721a3a5bba2148e7711b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/221320b71cfb0a5c9dfc06a175992b10fd572cdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/221320b71cfb0a5c9dfc06a175992b10fd572cdd"}]},"branch":"refs/heads/master"},"882e7fbb045d8b81e1a78e9a0763df0716695073":{"kind":"REWORK","_number":9,"created":"2021-05-26 16:48:12.000000000","uploader":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"ref":"refs/changes/47/792747/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/47/792747/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/47/792747/9"}}},"commit":{"parents":[{"commit":"04e5ead7c000211a4c10104ed2bb65c9df7681ae","subject":"Pass multipath config while creating connector object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/04e5ead7c000211a4c10104ed2bb65c9df7681ae"}]}],"author":{"name":"stress-t","email":"stress.t@gmail.com","date":"2021-05-23 15:11:15.000000000","tz":180},"committer":{"name":"stress_t","email":"stress.t@gmail.com","date":"2021-05-26 16:47:53.000000000","tz":180},"subject":"Allow to check the expected volume status multiple times before throwing an exception.","message":"Allow to check the expected volume status multiple times before throwing an exception.\n\nA case:\n- glance-api has been run two or more times on different servers with multiple threads.\n- using glance api in HA mode  and cinder as the backend.\n- we cannot use multiattach volumes.\n- Create a new virtual machine with temporary root disks. nova asks the glance api for the image, the glance api asks the cinder for the volume and the glance api try to mount the volume as a local disk to download nova. Cinder sends a response, and there we indicate a race condition for the two or more glance api servers. All answers from cinder include the volume status \"available\", but for the first glance of the api it is indeed \"available\" and for others it is \"in-use\".\n\nChange-Id: Ib281337e350ea38a2911721a3a5bba2148e7711b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/882e7fbb045d8b81e1a78e9a0763df0716695073"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/882e7fbb045d8b81e1a78e9a0763df0716695073"}]},"branch":"refs/heads/master"},"46ec47bf8410c91c80ffdfcdf44c82bbf665ac9a":{"kind":"REWORK","_number":10,"created":"2021-05-26 18:44:58.000000000","uploader":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"ref":"refs/changes/47/792747/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/47/792747/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/47/792747/10"}}},"commit":{"parents":[{"commit":"04e5ead7c000211a4c10104ed2bb65c9df7681ae","subject":"Pass multipath config while creating connector object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/04e5ead7c000211a4c10104ed2bb65c9df7681ae"}]}],"author":{"name":"stress-t","email":"stress.t@gmail.com","date":"2021-05-23 15:11:15.000000000","tz":180},"committer":{"name":"stress_t","email":"stress.t@gmail.com","date":"2021-05-26 18:44:43.000000000","tz":180},"subject":"Allow to check the expected volume status multiple times before throwing an exception.","message":"Allow to check the expected volume status multiple times before throwing an exception.\n\nA case:\n- glance-api has been run two or more times on different servers with multiple threads.\n- using glance api in HA mode  and cinder as the backend.\n- we cannot use multiattach volumes.\n- Create a new virtual machine with temporary root disks. nova asks the glance api for the image, the glance api asks the cinder for the volume and the glance api try to mount the volume as a local disk to download nova. Cinder sends a response, and there we indicate a race condition for the two or more glance api servers. All answers from cinder include the volume status \"available\", but for the first glance of the api it is indeed \"available\" and for others it is \"in-use\".\n\nChange-Id: Ib281337e350ea38a2911721a3a5bba2148e7711b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/46ec47bf8410c91c80ffdfcdf44c82bbf665ac9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/46ec47bf8410c91c80ffdfcdf44c82bbf665ac9a"}]},"branch":"refs/heads/master"},"0b200a53b16765e1083195e5f14b83c9553d4d28":{"kind":"REWORK","_number":11,"created":"2021-05-27 09:35:50.000000000","uploader":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"ref":"refs/changes/47/792747/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/47/792747/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/47/792747/11"}}},"commit":{"parents":[{"commit":"04e5ead7c000211a4c10104ed2bb65c9df7681ae","subject":"Pass multipath config while creating connector object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/04e5ead7c000211a4c10104ed2bb65c9df7681ae"}]}],"author":{"name":"stress-t","email":"stress.t@gmail.com","date":"2021-05-23 15:11:15.000000000","tz":180},"committer":{"name":"stress_t","email":"stress.t@gmail.com","date":"2021-05-27 09:35:36.000000000","tz":180},"subject":"Allow to check the expected volume status multiple times before throwing an exception.","message":"Allow to check the expected volume status multiple times before throwing an exception.\n\nA case:\n- glance-api has been run two or more times on different servers with multiple threads.\n- using glance api in HA mode  and cinder as the backend.\n- we cannot use multiattach volumes.\n- Create a new virtual machine with temporary root disks. nova asks the glance api for the image, the glance api asks the cinder for the volume and the glance api try to mount the volume as a local disk to download nova. Cinder sends a response, and there we indicate a race condition for the two or more glance api servers. All answers from cinder include the volume status \"available\", but for the first glance of the api it is indeed \"available\" and for others it is \"in-use\".\n\nChange-Id: Ib281337e350ea38a2911721a3a5bba2148e7711b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0b200a53b16765e1083195e5f14b83c9553d4d28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0b200a53b16765e1083195e5f14b83c9553d4d28"}]},"branch":"refs/heads/master"},"cf77e7ef53087a091c7e7bfce234f12e996e5e67":{"kind":"REWORK","_number":12,"created":"2021-05-27 09:38:24.000000000","uploader":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"ref":"refs/changes/47/792747/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/47/792747/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/47/792747/12"}}},"commit":{"parents":[{"commit":"04e5ead7c000211a4c10104ed2bb65c9df7681ae","subject":"Pass multipath config while creating connector object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/04e5ead7c000211a4c10104ed2bb65c9df7681ae"}]}],"author":{"name":"stress-t","email":"stress.t@gmail.com","date":"2021-05-23 15:11:15.000000000","tz":180},"committer":{"name":"stress_t","email":"stress.t@gmail.com","date":"2021-05-27 09:38:10.000000000","tz":180},"subject":"Allow to check the expected volume status multiple times before throwing an exception.","message":"Allow to check the expected volume status multiple times before throwing an exception.\n\nA case:\n- glance-api has been run two or more times on different servers with multiple threads.\n- using glance api in HA mode  and cinder as the backend.\n- we cannot use multiattach volumes.\n- Create a new virtual machine with temporary root disks. nova asks the glance api for the image, the glance api asks the cinder for the volume and the glance api try to mount the volume as a local disk to download nova. Cinder sends a response, and there we indicate a race condition for the two or more glance api servers. All answers from cinder include the volume status \"available\", but for the first glance of the api it is indeed \"available\" and for others it is \"in-use\".\n\nChange-Id: Ib281337e350ea38a2911721a3a5bba2148e7711b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/cf77e7ef53087a091c7e7bfce234f12e996e5e67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/cf77e7ef53087a091c7e7bfce234f12e996e5e67"}]},"branch":"refs/heads/master"},"3b80fb01ad985af68ed5c9f2803ad499dacff8fb":{"kind":"REWORK","_number":13,"created":"2021-05-27 11:56:51.000000000","uploader":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"ref":"refs/changes/47/792747/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/47/792747/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/47/792747/13"}}},"commit":{"parents":[{"commit":"04e5ead7c000211a4c10104ed2bb65c9df7681ae","subject":"Pass multipath config while creating connector object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/04e5ead7c000211a4c10104ed2bb65c9df7681ae"}]}],"author":{"name":"stress-t","email":"stress.t@gmail.com","date":"2021-05-23 15:11:15.000000000","tz":180},"committer":{"name":"stress_t","email":"stress.t@gmail.com","date":"2021-05-27 11:56:34.000000000","tz":180},"subject":"Allow to check the expected volume status multiple times before throwing an exception.","message":"Allow to check the expected volume status multiple times before throwing an exception.\n\nA case:\n- glance-api has been run two or more times on different servers with multiple threads.\n- using glance api in HA mode  and cinder as the backend.\n- we cannot use multiattach volumes.\n- Create a new virtual machine with temporary root disks. nova asks the glance api for the image, the glance api asks the cinder for the volume and the glance api try to mount the volume as a local disk to download nova. Cinder sends a response, and there we indicate a race condition for the two or more glance api servers. All answers from cinder include the volume status \"available\", but for the first glance of the api it is indeed \"available\" and for others it is \"in-use\".\n\nChange-Id: Ib281337e350ea38a2911721a3a5bba2148e7711b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3b80fb01ad985af68ed5c9f2803ad499dacff8fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3b80fb01ad985af68ed5c9f2803ad499dacff8fb"}]},"branch":"refs/heads/master"},"94c970faec576ce235cf4963d7987a9a145e7b04":{"kind":"REWORK","_number":14,"created":"2021-06-15 06:50:35.000000000","uploader":{"_account_id":32372,"name":"StreSS.t","email":"stress.t@gmail.com","username":"stress_t"},"ref":"refs/changes/47/792747/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/47/792747/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/47/792747/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/47/792747/14"}}},"commit":{"parents":[{"commit":"04e5ead7c000211a4c10104ed2bb65c9df7681ae","subject":"Pass multipath config while creating connector object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/04e5ead7c000211a4c10104ed2bb65c9df7681ae"}]}],"author":{"name":"stress-t","email":"stress.t@gmail.com","date":"2021-05-23 15:11:15.000000000","tz":180},"committer":{"name":"stress_t","email":"stress.t@gmail.com","date":"2021-06-15 06:49:16.000000000","tz":180},"subject":"Allow to check the expected volume status multiple times before throwing an exception.","message":"Allow to check the expected volume status multiple times before throwing an exception.\n\nA case:\n- glance-api has been run two or more times on different servers with multiple threads.\n- using glance api in HA mode  and cinder as the backend.\n- we cannot use multiattach volumes.\n- Create a new virtual machine with temporary root disks. nova asks the glance api for the image, the glance api asks the cinder for the volume and the glance api try to mount the volume as a local disk to download nova. Cinder sends a response, and there we indicate a race condition for the two or more glance api servers. All answers from cinder include the volume status \"available\", but for the first glance of the api it is indeed \"available\" and for others it is \"in-use\".\n\nChange-Id: Ib281337e350ea38a2911721a3a5bba2148e7711b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/94c970faec576ce235cf4963d7987a9a145e7b04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/94c970faec576ce235cf4963d7987a9a145e7b04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
