)]}'
{"id":"openstack%2Fglance_store~793546","triplet_id":"openstack%2Fglance_store~master~If844b485970c4d0e0b13dcf431fe0e5138004b6e","project":"openstack/glance_store","branch":"master","topic":"upper-contraints","attention_set":{},"removed_from_attention_set":{"31825":{"account":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"last_update":"2023-02-09 16:26:12.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-02-09 16:26:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If844b485970c4d0e0b13dcf431fe0e5138004b6e","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","status":"MERGED","created":"2021-05-28 07:52:10.000000000","updated":"2023-02-09 16:27:24.000000000","submitted":"2023-02-09 16:26:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"793546-upper-contraints","meta_rev_id":"e1c5aa0593eef922aaad60a624adaca44ecfcd8e","_number":793546,"virtual_id_number":793546,"owner":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-09 16:26:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2023-02-09 14:28:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2023-02-09 14:25:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2023-02-09 14:28:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-28 09:14:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-27 01:58:38.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-07-07 21:04:42.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2022-07-07 22:56:08.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-02-09 14:25:05.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"e2a587db7e388edcc46719030cae126d88dedad3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"date":"2021-05-28 07:52:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e047c500baf8ef5ebc78595a9e7cb4c3a22dd81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 09:14:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be0d45d7ff614188a610180e344720db : SUCCESS in 1h 21m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da794bc8ff62472c8673fba564dcb841 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d93c0905f228464c93af7f910e700c84 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ccef803a4b7462a922f977005adf95d : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f958b34ab672487e83bb668abb118503 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a5189f0245c49ae9625e5d2ec438fa9 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88d82342eb7c48fa88e9167a4ea7cd79 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/367c7bb653fd406787aa7741717dc8a3 : SUCCESS in 5m 14s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/6aae72e979aa488eba2df3caf679dd1b : SUCCESS in 12m 53s","accounts_in_message":[],"_revision_number":1},{"id":"140ac4a21d2cbc0731fdf0d1d5cdad27e1a636da","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-10-27 01:58:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d660b8651c75c4a72974b12e318f23acabc126cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-01-12 22:28:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6c0672e8826391f264261c39471f30af2cacb0e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-01-12 22:29:36.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"000ff3508d73b5e29b737f7d0106abb2911bc30b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 23:53:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41e4067df18b41f0b1279f47f0ca0670 : SUCCESS in 1h 22m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ebf2f0b27374c07992b9fa04adc82dc : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ca4b5eb15c440778c6a510fd3ac2900 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b4b63c32f8a495b994b698ce6bf8c48 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a47e9412e98f40eba5d9c38c5576f604 : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1e32e1f8b4244d5b5fe755de912469d : SUCCESS in 3m 41s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/756a9ee0814c44e682937ac575cdd2c4 : SUCCESS in 8m 58s","accounts_in_message":[],"_revision_number":3},{"id":"09d20bd0c5420a491e6c2a76e72cef010af2ab28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-07 14:54:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"71b235466c3fdbec54b3095a55f24ef8b45e7087","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 16:14:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8820bb88524d43ed8bea0be777265cbc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/553ff47ac23046c3a9efe3b5acdbf2b4 : SUCCESS in 1h 18m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3744f93c62cd4423a79dfb31cb2662b2 : FAILURE in 2m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/299c34b95d1243ecbb2934ef990c284c : FAILURE in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df662cd055fa412c97df018345f501da : FAILURE in 2m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83bc096b55724549b4b7c4a539acf46a : FAILURE in 2m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b7241547459435f84e0d65321e528b5 : FAILURE in 2m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce62c3df86114861b023a515b618eaaa : SUCCESS in 3m 01s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/c36a99e87ddb4611aef136f5f2b271be : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":4},{"id":"5f1598db9f0959ef5257f88e6fba98d06240242b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-07-07 21:04:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1f86748f197dd77487903359768fffe851c8466c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 22:22:53.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/faea6d8939ef43a593ee0fb5bb37c2f5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b70609674844faab98db5e7e2f5a542 : SUCCESS in 1h 16m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73345196d07b4dd18a2f041ed05136c8 : FAILURE in 2m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f8733d186fc4f3b9cb4b98f17663f40 : FAILURE in 2m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f0e817659dd4e5cada62965c4abc0da : FAILURE in 2m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/482924e0b1aa4b4c8a485397c30da21a : FAILURE in 2m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0484383026a943c7bd8671d464c659a1 : FAILURE in 2m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0354e5b809784119b377bdc8b2ccb89c : SUCCESS in 2m 47s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/382ae587ced448799b92e6e27ef17468 : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":4},{"id":"8d3ccbfdd931dc5f61c410bdc302de1bd81b2d8d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-07-07 22:55:03.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"81e185dd261bd7ec4ee0cf45416c94b4540e90f4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-07-07 22:56:08.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8ae24d8762beba2729bb631cac4e08aeae64a210","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-07-07 23:00:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"767cb4a9cd2ef54067897fbff5191c7b646711cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-08-05 15:00:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1c6ac67da9380241f3fa2647c0786f06c4ffce75","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-08-05 15:03:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"680437a2177ccb25d03ec02d094285844a9ec8b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 16:34:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/01bce728bccd4398a537048587f3e003\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50e3fcb622274cffb6f6886d6308b18f : SUCCESS in 1h 32m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d8788e8decf46b696f2ea054ad8ad16 : FAILURE in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/657b64e9e9144ee1984aca2f12af8081 : FAILURE in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bb6ed54774f41608968b3e98f133ec6 : FAILURE in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9348eb3cb2c54ba7b624e18bae67c17f : FAILURE in 3m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54558758c77446748802cf1afed771e7 : FAILURE in 3m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfd918a6d8124cab97b288e9ed1112e2 : SUCCESS in 3m 46s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/52aa4f101b6a4897860d85f7ed25e15f : SUCCESS in 8m 16s","accounts_in_message":[],"_revision_number":5},{"id":"faedb0f280e83b91f527dfe440e538bf30ad841e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-06 19:13:18.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"4ab2f1a0d49fe4bb7961a5ff81b3ed21cbb07ffa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 21:33:56.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c348ee9450ac4adc89994378ddd4fde7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d756a6252cc2472787a7fbbb33938a48 : TIMED_OUT in 2h 04m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/457b5311d15c4ec588f5ebe8f54afd4e : FAILURE in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d9ee6fdd71f476790781d889090b352 : FAILURE in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/039be706b6e444bf8ac3fa38ea7be952 : FAILURE in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/444d610f44594589afbedd4cac2c5e81 : FAILURE in 2m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0879fa48da7941e2b3171955772235c0 : SUCCESS in 3m 26s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/0fa2a19c8f7c417a87ae76985cc3acf9 : SUCCESS in 8m 31s","accounts_in_message":[],"_revision_number":5},{"id":"7e936ba7304c746cc834bd9ccb89044d0ea29a53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-06 21:41:47.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"4967ea5a21ae6c9d42270a3bb47ce68f1acf434f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-06 21:43:49.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"dc7c60b10eaac6d9a9a2d8401b48a6cdbe01ba45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 23:36:45.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4f1e92eed0c43599762775adbdd7bb1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/700f3b39f4084c8ebf52a4fbf2df5280 : SUCCESS in 1h 47m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70909884da3f4020a3da6b0e33e1122a : FAILURE in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f30067719004730a0b9dfc229f76817 : FAILURE in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b8b48eca8ed475894d201b850b1cf9a : FAILURE in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b6836729c7e450fb7e9d83e74cb71dc : FAILURE in 2m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73ab93d4447c4c149a1edaa73b3477e8 : SUCCESS in 3m 09s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/a8ef6067480340adbf2fd959ba8c30eb : SUCCESS in 7m 54s","accounts_in_message":[],"_revision_number":6},{"id":"fb9ac77ff5d866f22841a1d3dd3542688ef8e2c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-07 01:48:59.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7634100a14207484d4082db6b43263517d294189","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 03:41:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30b115221df24f5585c6445464a4c38f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/944c384e98a447e3ab6490f223e83cd2 : SUCCESS in 1h 45m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59c58018879c441aa99129eb48fda559 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c1ae30a527245f6a70a816bd3b85337 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/122e7fb329a1464aa5092b84ed3157f4 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79fbe1335a814663b228df286ddd98df : SUCCESS in 2m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b748aec57bd42f5aa14e3705b5308a6 : SUCCESS in 2m 41s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/120fd59e53654170bd2ca4a4cbc583aa : SUCCESS in 13m 30s","accounts_in_message":[],"_revision_number":7},{"id":"78dd690f927b236cf2e90301ee8456a539c3a38d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-09 14:25:05.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0dea8c3223cf90b62daf97d1d8218898ea28296a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-02-09 14:28:39.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"73988e4f0ef03172bc38c7a58874cc76ff89b809","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 14:29:19.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"e41bce59d52b361bf82b41e7c8715e37414e8f57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 16:26:12.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c02e9f6a5cfc4130b67680b477a560ae\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6abebc35c7194142b213a15dc58a682e : SUCCESS in 1h 50m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5808ac267df49d79e80174908346b35 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5511d7ef954645fe9c7aef75ed0a2442 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/268f6c727db14a57b9eabe356670c11f : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23467512b50e40df965bd4624ef2402b : SUCCESS in 3m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bc421287afc4248880218c8d9094fce : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":7},{"id":"06703cbf178f0350450be2aa9feff75ec9dffc7c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 16:26:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"e1c5aa0593eef922aaad60a624adaca44ecfcd8e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 16:27:24.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60148abca7564c9ab5869538b37bbf74\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d86c5f40ea3f4d12b72a0f28ff768ea5 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/094ca1ab933b4e9184d8a695640e3a2e : SUCCESS in 47s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"6b49171e024ca3a69a77706ae7d2fe124d5aad7d","revisions":{"795938f245aae744662c185b7f50415d8897b39d":{"kind":"REWORK","_number":1,"created":"2021-05-28 07:52:10.000000000","uploader":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"ref":"refs/changes/46/793546/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/46/793546/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/46/793546/1"}}},"commit":{"parents":[{"commit":"057b6e5305d7059db28c9d6c0319a38a43fa7acf","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/057b6e5305d7059db28c9d6c0319a38a43fa7acf"}]}],"author":{"name":"XinxinShen","email":"shenxinxin@inspur.com","date":"2021-05-28 07:50:44.000000000","tz":480},"committer":{"name":"XinxinShen","email":"shenxinxin@inspur.com","date":"2021-05-28 07:50:59.000000000","tz":480},"subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","message":"Replace deprecated UPPER_CONSTRAINTS_FILE variable\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\n\n[1] https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\n\nChange-Id: If844b485970c4d0e0b13dcf431fe0e5138004b6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/795938f245aae744662c185b7f50415d8897b39d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/795938f245aae744662c185b7f50415d8897b39d"}]},"branch":"refs/heads/master"},"ed038a5f9d14be8e30da7a4570c93af64d476e2e":{"kind":"REWORK","_number":2,"created":"2022-01-12 22:28:57.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/46/793546/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/46/793546/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/46/793546/2"}}},"commit":{"parents":[{"commit":"16b9d1f7adc503c4ed1108135a4ef4437abf7fa5","subject":"Merge \"Correct attachment_complete call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/16b9d1f7adc503c4ed1108135a4ef4437abf7fa5"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-01-12 22:27:43.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-01-12 22:27:43.000000000","tz":60},"subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","message":"Replace deprecated UPPER_CONSTRAINTS_FILE variable\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\n\nSee https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: If844b485970c4d0e0b13dcf431fe0e5138004b6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ed038a5f9d14be8e30da7a4570c93af64d476e2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ed038a5f9d14be8e30da7a4570c93af64d476e2e"}]},"branch":"refs/heads/master"},"2fe121f6e3193a843042d001672e00c0725ee359":{"kind":"NO_CHANGE","_number":3,"created":"2022-01-12 22:29:36.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/46/793546/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/46/793546/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/46/793546/3"}}},"commit":{"parents":[{"commit":"16b9d1f7adc503c4ed1108135a4ef4437abf7fa5","subject":"Merge \"Correct attachment_complete call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/16b9d1f7adc503c4ed1108135a4ef4437abf7fa5"}]}],"author":{"name":"Xinxin Shen","email":"shenxinxin@inspur.com","date":"2022-01-12 22:27:43.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-01-12 22:29:15.000000000","tz":60},"subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","message":"Replace deprecated UPPER_CONSTRAINTS_FILE variable\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\n\nSee https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: If844b485970c4d0e0b13dcf431fe0e5138004b6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2fe121f6e3193a843042d001672e00c0725ee359"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2fe121f6e3193a843042d001672e00c0725ee359"}]},"branch":"refs/heads/master"},"3968fee396644f6b4f8ca6ad6f38e2519c2404ba":{"kind":"REWORK","_number":4,"created":"2022-07-07 14:54:35.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/46/793546/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/46/793546/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/46/793546/4"}}},"commit":{"parents":[{"commit":"16b9d1f7adc503c4ed1108135a4ef4437abf7fa5","subject":"Merge \"Correct attachment_complete call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/16b9d1f7adc503c4ed1108135a4ef4437abf7fa5"}]}],"author":{"name":"Xinxin Shen","email":"shenxinxin@inspur.com","date":"2022-01-12 22:27:43.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-07-07 14:53:01.000000000","tz":120},"subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","message":"Replace deprecated UPPER_CONSTRAINTS_FILE variable\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\n\nSee https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: If844b485970c4d0e0b13dcf431fe0e5138004b6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3968fee396644f6b4f8ca6ad6f38e2519c2404ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3968fee396644f6b4f8ca6ad6f38e2519c2404ba"}]},"branch":"refs/heads/master"},"ad5ba82a5a43ae57b6d0cb0c1909d82665c6f058":{"kind":"REWORK","_number":5,"created":"2022-08-05 15:00:34.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/46/793546/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/46/793546/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/46/793546/5"}}},"commit":{"parents":[{"commit":"6851cab51a3d5dbf018f755efee8e0e640ac12ab","subject":"Add debug logs to cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6851cab51a3d5dbf018f755efee8e0e640ac12ab"}]}],"author":{"name":"Xinxin Shen","email":"shenxinxin@inspur.com","date":"2022-01-12 22:27:43.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-08-05 14:42:32.000000000","tz":120},"subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","message":"Replace deprecated UPPER_CONSTRAINTS_FILE variable\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\n\nSee https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: If844b485970c4d0e0b13dcf431fe0e5138004b6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ad5ba82a5a43ae57b6d0cb0c1909d82665c6f058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ad5ba82a5a43ae57b6d0cb0c1909d82665c6f058"}]},"branch":"refs/heads/master"},"08e21d1d103ccd8371d3ccf3fe292f120d238a33":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-02-06 21:41:47.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/46/793546/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/46/793546/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/46/793546/6"}}},"commit":{"parents":[{"commit":"2379a56546be403d1b3fc62947538c0548899871","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2379a56546be403d1b3fc62947538c0548899871"}]}],"author":{"name":"Xinxin Shen","email":"shenxinxin@inspur.com","date":"2022-01-12 22:27:43.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-02-06 21:41:47.000000000","tz":0},"subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","message":"Replace deprecated UPPER_CONSTRAINTS_FILE variable\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\n\nSee https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: If844b485970c4d0e0b13dcf431fe0e5138004b6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/08e21d1d103ccd8371d3ccf3fe292f120d238a33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/08e21d1d103ccd8371d3ccf3fe292f120d238a33"}]},"branch":"refs/heads/master","description":"Rebase"},"6b49171e024ca3a69a77706ae7d2fe124d5aad7d":{"kind":"REWORK","_number":7,"created":"2023-02-07 01:48:59.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/46/793546/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/46/793546/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/793546/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/46/793546/7"}}},"commit":{"parents":[{"commit":"2379a56546be403d1b3fc62947538c0548899871","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2379a56546be403d1b3fc62947538c0548899871"}]}],"author":{"name":"Xinxin Shen","email":"shenxinxin@inspur.com","date":"2022-01-12 22:27:43.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-02-07 01:48:42.000000000","tz":60},"subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","message":"Replace deprecated UPPER_CONSTRAINTS_FILE variable\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\n\nSee https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: If844b485970c4d0e0b13dcf431fe0e5138004b6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6b49171e024ca3a69a77706ae7d2fe124d5aad7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6b49171e024ca3a69a77706ae7d2fe124d5aad7d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
