)]}'
{"id":"openstack%2Fglance_store~800183","triplet_id":"openstack%2Fglance_store~stable%2Fwallaby~Icecac80dd9e4e9f7ffa76bb7ca63d8d112036b70","project":"openstack/glance_store","branch":"stable/wallaby","topic":"bug/1934849-stable/wallaby","hashtags":[],"change_id":"Icecac80dd9e4e9f7ffa76bb7ca63d8d112036b70","subject":"s3: Optimize WRITE_CHUNKSIZE to minimize an overhead","status":"ABANDONED","created":"2021-07-10 07:51:40.000000000","updated":"2022-05-05 15:06:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":799870,"cherry_pick_of_patch_set":3,"meta_rev_id":"8da7ee20f6d5e14faaac17fb747efba314f62a4e","_number":800183,"virtual_id_number":800183,"owner":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-07-10 09:34:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-10 09:34:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4fe693c9181c24df422f563519bb0a1bc438288c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2021-07-10 07:51:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cb1b519ad90fec4dfda2e87cefef732da7c3a36a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 09:34:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f21868925b4f4ae0998f89837fc0bd4d : SUCCESS in 1h 24m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75aacd735fd0452294afa099f756f347 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acf62a5700c240a4bf00735fa5c612d5 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20142e439a7344ce9981f93ba6bfa0fe : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0f4e389e262483a8a555bd522fe7939 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60b91a62d862450bb7d43cc60d328c16 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/525b21a1a0894323aefd2a80986433ec : SUCCESS in 4m 33s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/4ad1a08821d141aaab11c307bfe7adb6 : SUCCESS in 9m 16s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/4cbaa7ce83c04e829502373433aa6c0b : SUCCESS in 1h 40m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8da7ee20f6d5e14faaac17fb747efba314f62a4e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-05 15:06:20.000000000","message":"Abandoned\n\nThis does not fits with upstream stable branch policy (current branch - 2), feel free to reach back to glance team on #openstack-glance irc channel","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb03f56094419e1b89e52a790e003a08277fa527","revisions":{"eb03f56094419e1b89e52a790e003a08277fa527":{"kind":"REWORK","_number":1,"created":"2021-07-10 07:51:40.000000000","uploader":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"ref":"refs/changes/83/800183/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/83/800183/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/83/800183/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/83/800183/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/83/800183/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/83/800183/1"}}},"commit":{"parents":[{"commit":"9bb857097e8847bf8c9c722f7ff5990dccb391cf","subject":"Pass multipath config while creating connector object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/9bb857097e8847bf8c9c722f7ff5990dccb391cf"}]}],"author":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2021-07-07 16:36:38.000000000","tz":540},"committer":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2021-07-10 07:51:40.000000000","tz":0},"subject":"s3: Optimize WRITE_CHUNKSIZE to minimize an overhead","message":"s3: Optimize WRITE_CHUNKSIZE to minimize an overhead\n\nWhen processing the image data from a client, the previous and small\nchunk size had a significant overhead. Let\u0027s be aligned with other chunk\nsize related parameters such as DEFAULT_LARGE_OBJECT_MIN_CHUNK_SIZE\n(\u003d5MiB) and DEFAULT_LARGE_OBJECT_CHUNK_SIZE (\u003d10MiB).\n\nThe performance difference should be tangible especially with a\nsinglepart upload. And the upload time can be shortened for example as\nfollows:\n\n100 MiB:   36.5s -\u003e 4.0s\n200 MiB: 2m16.8s -\u003e 6.1s\n300 MiB: 4m50.4s -\u003e 9.1s\n\nCloses-Bug: 1934849\nChange-Id: Icecac80dd9e4e9f7ffa76bb7ca63d8d112036b70\n(cherry picked from commit 32f9a1509bc94baa6acaf508c922f3b7edf5b65f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/eb03f56094419e1b89e52a790e003a08277fa527"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/eb03f56094419e1b89e52a790e003a08277fa527"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
