)]}'
{"id":"openstack%2Fglance_store~804346","triplet_id":"openstack%2Fglance_store~stable%2Fwallaby~I1de76a70e07bbf2bd70900b4d93104a1b0fda74b","project":"openstack/glance_store","branch":"stable/wallaby","topic":"804346","hashtags":[],"change_id":"I1de76a70e07bbf2bd70900b4d93104a1b0fda74b","subject":"Drop lower-constraints job","status":"MERGED","created":"2021-08-12 14:53:28.000000000","updated":"2021-08-12 18:35:15.000000000","submitted":"2021-08-12 18:32:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804346-804346","cherry_pick_of_change":804369,"cherry_pick_of_patch_set":1,"meta_rev_id":"3c49291b34793ec8d53c4739325837cdaa684e90","_number":804346,"virtual_id_number":804346,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-12 18:32:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-08-12 16:35:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-08-12 17:01:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-08-12 16:35:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-12 14:54:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-12 15:05:45.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-08-12 15:05:45.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"e9f79c178ec21382b668165de696da9b0ea5b9a4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-12 14:53:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* .zuul.yaml\n* lower-constraints.txt","accounts_in_message":[],"_revision_number":1},{"id":"468e44f7dda4dd99a2cfb1bfdd0f45615c4416ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 14:54:31.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance_store on branch stable/wallaby.  The error was:\n\n  while scanning a simple key\n  in \"openstack/glance_store/.zuul.yaml@stable/wallaby\", line 159, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/glance_store/.zuul.yaml@stable/wallaby\", line 162, column 1","accounts_in_message":[],"_revision_number":1},{"id":"2463fbb2756d8a3344282c05b1d0f06ec2e07b6b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-12 15:01:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c37e39d0d1cba886b87d6c162f953bdc8e64ac39","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-12 15:05:45.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"df999a3cd0ac20a2ebb8a4ec31b40bfa1c6a3c04","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-12 16:35:51.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"a6ab37f35a14f485976bba7f34a8bb215e3ca83f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 16:54:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e055c7fc59cb4f1b91f6884414a7dfb8 : SUCCESS in 2m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93583d18ab06403ab7032c805af888fc : SUCCESS in 1h 31m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61cbf6c81a60465f8908f979dcec1876 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/732ccac50dd44dfa8b180f25117c7fd2 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7964a26138da40d5a4bafff5f916b9d2 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48976ea050574f8eb54b1062bff394ff : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14ffe85a3a4646a79084eb9f8ca297b8 : SUCCESS in 4m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cbe9fda62c3482dbb0f69a0170ce148 : SUCCESS in 4m 10s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/6c63300ff02b412cac43e3ae4ef03504 : SUCCESS in 8m 42s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/ed7a9c1e29354afab40fe7c1279512ca : SUCCESS in 1h 48m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0da31a80f1451fe898445c2be3fa79a4e6750811","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-12 17:01:28.000000000","message":"Patch Set 2: Workflow+1\n\nNinja approving as it is straight forward and need to backport to victoria branch as well!","accounts_in_message":[],"_revision_number":2},{"id":"fa92d53b71d8aeeecb47a4f8970746d34ece3344","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 17:02:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"43e708ba64883c3eda35c1ba874218d4804a2075","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 18:32:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/25a630d523e34316ab28fa63dc3b5f40 : SUCCESS in 2m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b6f6777c01a4a8aa0ddc0ff887d3906 : SUCCESS in 1h 28m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31afe18546fd4122924e80b26cb40cea : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75bb609019e04bb9b6e8c529a191f6c4 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75d3669defaa481aabb0186069bddc11 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/763fcc7b11f749bab976f0b368b5b54b : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7384e77441945519f4df6bcbe08dec7 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":2},{"id":"4d09992cc3f253eeb514928b91de6757b833b679","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 18:32:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3c49291b34793ec8d53c4739325837cdaa684e90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 18:35:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b699641b5749409cb7c3619fba085941 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48c0d0b86fa14cbea2faf74c7ea8eb92 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b93facf1d71dccb9d3e4ca062d8362327330e263","revisions":{"9cae0a619b3792007556e6f7748120f0a09d121c":{"kind":"REWORK","_number":1,"created":"2021-08-12 14:53:28.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/46/804346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/46/804346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/804346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/804346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/804346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/46/804346/1"}}},"commit":{"parents":[{"commit":"9bb857097e8847bf8c9c722f7ff5990dccb391cf","subject":"Pass multipath config while creating connector object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/9bb857097e8847bf8c9c722f7ff5990dccb391cf"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-08-12 09:15:34.000000000","tz":-240},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-08-12 14:53:28.000000000","tz":0},"subject":"Drop lower-constraints job","message":"Drop lower-constraints job\n\nChange-Id: I1de76a70e07bbf2bd70900b4d93104a1b0fda74b\n(cherry picked from commit b42ffa40351b31fc93196b34932bb36056ff33db)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/9cae0a619b3792007556e6f7748120f0a09d121c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/9cae0a619b3792007556e6f7748120f0a09d121c"}]},"branch":"refs/heads/stable/wallaby"},"b93facf1d71dccb9d3e4ca062d8362327330e263":{"kind":"REWORK","_number":2,"created":"2021-08-12 15:01:28.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/46/804346/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/46/804346/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/804346/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/804346/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/46/804346/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/46/804346/2"}}},"commit":{"parents":[{"commit":"9bb857097e8847bf8c9c722f7ff5990dccb391cf","subject":"Pass multipath config while creating connector object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/9bb857097e8847bf8c9c722f7ff5990dccb391cf"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-08-12 09:15:34.000000000","tz":-240},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-08-12 15:01:24.000000000","tz":-240},"subject":"Drop lower-constraints job","message":"Drop lower-constraints job\n\nChange-Id: I1de76a70e07bbf2bd70900b4d93104a1b0fda74b\n(cherry picked from commit b42ffa40351b31fc93196b34932bb36056ff33db)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/b93facf1d71dccb9d3e4ca062d8362327330e263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/b93facf1d71dccb9d3e4ca062d8362327330e263"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
