)]}'
{"id":"openstack%2Fglance_store~804606","triplet_id":"openstack%2Fglance_store~stable%2Fvictoria~I1de76a70e07bbf2bd70900b4d93104a1b0fda74b","project":"openstack/glance_store","branch":"stable/victoria","hashtags":[],"change_id":"I1de76a70e07bbf2bd70900b4d93104a1b0fda74b","subject":"Drop lower-constraints job","status":"MERGED","created":"2021-08-16 07:42:09.000000000","updated":"2021-08-16 17:08:27.000000000","submitted":"2021-08-16 17:08:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804606","meta_rev_id":"74874e8933c94aeea185ce923dc6bf4e232f6b7e","_number":804606,"virtual_id_number":804606,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-16 17:08:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-08-16 10:26:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-08-16 10:26:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-16 07:44:55.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2021-08-16 07:44:55.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-08-16 07:44:55.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-08-16 09:17:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1592b735d4daeef1efbf7009281a5824b286080d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-16 07:42:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28229a5e1537614e84001e183cf041e96155bfe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 09:17:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/58b46b1c56114cffb2ac0b6ae36aa740 : SUCCESS in 2m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea4ae4400558466ca996c7356ae53865 : SUCCESS in 1h 14m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/573e298ffffe4e64b82989bcecf0d1a7 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0537a0241d64e1c81cdc81ee29922f7 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e03ac7468ade4242a073ce22841d5d0d : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f413793d01c4650b8affd9abc685424 : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88efeef255f0434ea367ab9d2453eed9 : SUCCESS in 4m 04s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/0a1ff046e03440aeb247e1a16301f097 : FAILURE in 1h 33m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e8e46faae176caeb8d3fb691b1ebe94c0c39e36","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-16 09:27:15.000000000","message":"Patch Set 1:\n\nIdeally I think this should be cherry-picked from stable/wallaby and not master, right?","accounts_in_message":[],"_revision_number":1},{"id":"48263ddecfbfd310b604145fd48990ef099a9642","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-16 09:54:47.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Ideally I think this should be cherry-picked from stable/wallaby and not master, right?\n\nI did cherry pick it from wallaby but i had to resolve the merge conflict locally so i think in that case, gerrit doesn\u0027t include the \"cherry picked from commit\" statement and used the same one as wallaby patch\nWill update the commit message","accounts_in_message":[],"_revision_number":1},{"id":"4c85a07a4dd6aff51a9533fb6242103ae419a94f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-16 09:59:28.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"575e8a104d4da04b636c8b56459fca68e5b37183","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-16 10:00:00.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e Ideally I think this should be cherry-picked from stable/wallaby and not master, right?\n\u003e \n\u003e I did cherry pick it from wallaby but i had to resolve the merge conflict locally so i think in that case, gerrit doesn\u0027t include the \"cherry picked from commit\" statement and used the same one as wallaby patch\n\u003e Will update the commit message\n\nDone","accounts_in_message":[],"_revision_number":2},{"id":"9576c9a135a0abe6109a2e289ff5f4f46f1e68c1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-16 10:26:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLooks good now, clean and required backport!\nNinja approving it as it us just removing a job which no longer advisable to use.","accounts_in_message":[],"_revision_number":2},{"id":"e7f8bebb87dcd20568a4a860f4dc73748fb47333","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 12:07:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a92927e0800944569b7f398418d48891 : SUCCESS in 2m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ddfe46cd90634101a2e43d3c045dfdb1 : SUCCESS in 1h 27m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94f9e50acd4448028aef0ce0718fceb8 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1288c3aa5fa649db9f3d04fcca7386cd : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2558d691d42944b19ffbe51ac8799351 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b99437319bc4ecc8db56f8838caf73a : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49c6e89d31e44f81a658f251841151bc : SUCCESS in 4m 39s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/541b2e1121dc47d787127d4f0305b3cb : TIMED_OUT in 2h 07m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3685bfc44607fab28d30bb6ff333b073cfa7933b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 12:08:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6828515bd4310f12f38b94307d71aa4829f46abb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 13:38:56.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/86879c99ef90440aaa5478ef093063d8 : SUCCESS in 2m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e20a70040b884f139ee5128d75856b10 : FAILURE in 1h 26m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04551ea9f60c4b88adc250b36276ea37 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7426c06e0c6d47a6b4f77252dfce3de9 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a886975593d141c49a8ee456cd63cfa2 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b94aa13a2ba4acc993af298e22fa11f : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0df534fc919b4cc19f9b4fc916d966d7 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":2},{"id":"6ec0527f17d0864763250b96aa8283560be7505a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-16 13:54:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f11ab834fe324b23aedd8b05b3da6ce27372baf5","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-16 15:18:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4a86d52f81e6104c451da97d370dccae103c2631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 15:46:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f433f7c6ce204990a9cb15246e2d1aa2 : SUCCESS in 3m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09ee5b535cb241f3b773c64d5906f8fd : SUCCESS in 1h 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d3f2367e33b49bcba22d1d83ba4d652 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02baa0fe1cc84a17a9ecfe5296a2f679 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9d10768f2714289aa36c67188dc1c16 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/265e11df05214651acbf5c83ba0743b2 : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/108f056208c04388858e20b22f08d31d : SUCCESS in 4m 03s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/3add7d4a2abd4b0b8cd0144bef7ba996 : SUCCESS in 1h 47m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f998c1dee753aab486ffe9b2108b3e2fa49be0bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 15:46:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cda7481f7e08e6e4808ced398ae8010c796abf9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 17:08:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d3892fb31d2f4c3e8f5fdbe737e57e30 : SUCCESS in 2m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc07f91b76cb421fbb1b12ee7802557e : SUCCESS in 1h 17m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e5657afd6164ea1ae5138b67ad586e4 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/151c93456b794230aefe71db916c17e0 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5a1fd3be2b44f9284269bf5ef102589 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a08f8be4d54d4920ae65ea662589e39b : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ecdec8cca854a2182d5d72403154e89 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":2},{"id":"74874e8933c94aeea185ce923dc6bf4e232f6b7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 17:08:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ff8df6dc0e56e5c8305176140eb783bd9765b695","revisions":{"d1e07c303f26edf93379757ce15977a3dbabc6fc":{"kind":"REWORK","_number":1,"created":"2021-08-16 07:42:09.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/06/804606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/06/804606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/06/804606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/06/804606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/06/804606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/06/804606/1"}}},"commit":{"parents":[{"commit":"177b8034e34b3b556a536a29f9c61026a3532c25","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/177b8034e34b3b556a536a29f9c61026a3532c25"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-08-12 09:15:34.000000000","tz":-240},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-08-16 07:42:05.000000000","tz":-240},"subject":"Drop lower-constraints job","message":"Drop lower-constraints job\n\nChange-Id: I1de76a70e07bbf2bd70900b4d93104a1b0fda74b\n(cherry picked from commit b42ffa40351b31fc93196b34932bb36056ff33db)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d1e07c303f26edf93379757ce15977a3dbabc6fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d1e07c303f26edf93379757ce15977a3dbabc6fc"}]},"branch":"refs/heads/stable/victoria"},"ff8df6dc0e56e5c8305176140eb783bd9765b695":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-08-16 09:59:28.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/06/804606/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/06/804606/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/06/804606/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/06/804606/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/06/804606/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/06/804606/2"}}},"commit":{"parents":[{"commit":"177b8034e34b3b556a536a29f9c61026a3532c25","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/177b8034e34b3b556a536a29f9c61026a3532c25"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-08-12 09:15:34.000000000","tz":-240},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-08-16 09:55:20.000000000","tz":0},"subject":"Drop lower-constraints job","message":"Drop lower-constraints job\n\nChange-Id: I1de76a70e07bbf2bd70900b4d93104a1b0fda74b\n(cherry picked from commit b93facf1d71dccb9d3e4ca062d8362327330e263)\n(cherry picked from commit b42ffa40351b31fc93196b34932bb36056ff33db)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ff8df6dc0e56e5c8305176140eb783bd9765b695"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ff8df6dc0e56e5c8305176140eb783bd9765b695"}]},"branch":"refs/heads/stable/victoria","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
