)]}'
{"id":"openstack%2Fglance_store~804822","triplet_id":"openstack%2Fglance_store~stable%2Fwallaby~I8403a65a54adf375d2e8c309d710c12e4f34661f","project":"openstack/glance_store","branch":"stable/wallaby","hashtags":[],"change_id":"I8403a65a54adf375d2e8c309d710c12e4f34661f","subject":"DNM: check lower constraints job removal","status":"ABANDONED","created":"2021-08-17 06:21:18.000000000","updated":"2021-12-10 13:39:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"017f874c62acd6b37d81422e114a99c33a7dc9d7","_number":804822,"virtual_id_number":804822,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-17 08:53:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-17 08:53:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5dfb9f5214f082f067ca96dfce61eb6ccf9404cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-17 06:21:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d084229029fb70913823c0967020cd8a100aefe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 08:53:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f64acf5d3614e09866edaf19ad6f120 : SUCCESS in 1h 55m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0270fb8ff6e4f18a37934448a401536 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b763a1df32845e29321ddb921a3bbea : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4045224e9f8545b88430358d8070b24c : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76e2845d7819412092561b9c3459dba4 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8f1c4727bdc41afb26bea34141671ae : SUCCESS in 4m 19s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/bfe9f301ec244022841d0c4bbb202b61 : SUCCESS in 9m 19s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/4da3d2a6fe1940fca7f7c702e6ad94a8 : FAILURE in 1h 36m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"017f874c62acd6b37d81422e114a99c33a7dc9d7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-12-10 13:39:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7bde266fbe558506d50dab359b435d294693952","revisions":{"a7bde266fbe558506d50dab359b435d294693952":{"kind":"REWORK","_number":1,"created":"2021-08-17 06:21:18.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/22/804822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/22/804822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/22/804822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/22/804822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/22/804822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/22/804822/1"}}},"commit":{"parents":[{"commit":"b93facf1d71dccb9d3e4ca062d8362327330e263","subject":"Drop lower-constraints job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/b93facf1d71dccb9d3e4ca062d8362327330e263"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-08-17 06:20:44.000000000","tz":-240},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-08-17 06:20:56.000000000","tz":-240},"subject":"DNM: check lower constraints job removal","message":"DNM: check lower constraints job removal\n\nChange-Id: I8403a65a54adf375d2e8c309d710c12e4f34661f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a7bde266fbe558506d50dab359b435d294693952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a7bde266fbe558506d50dab359b435d294693952"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
