)]}'
{"id":"openstack%2Fglance_store~804833","triplet_id":"openstack%2Fglance_store~stable%2Fvictoria~Ifaa4d39047a183c48a31ccb41e22534e1fc654a4","project":"openstack/glance_store","branch":"stable/victoria","topic":"dnm","hashtags":[],"change_id":"Ifaa4d39047a183c48a31ccb41e22534e1fc654a4","subject":"[DNM] Test lower constraints trigger","status":"ABANDONED","created":"2021-08-17 08:27:25.000000000","updated":"2021-08-19 10:58:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dda7ddf6849b046a509b8adad5df5dc5c3643c8e","_number":804833,"virtual_id_number":804833,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-17 10:16:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-17 08:27:25.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-08-17 10:16:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"178bac2661ee47b129dc71baee2c1b5d037df15c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-08-17 08:27:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c071bc9a11797335f6bd22009a70ee5a4423b38","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-08-17 08:40:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"07060b7426ebbee030f46fc9dc8c4f04ba739ee2","tag":"autogenerated:gerrit:restore","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-08-17 08:40:57.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"a5044c270af0a3f343933014bd4eb9aff7a280a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-08-17 08:43:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1f015613f380b2296f8964c0ad7592a468c1a218","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-08-17 08:47:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4e97da2de087ae5fb4700eb13f45b48d644ac702","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 10:16:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9298431aa574fe8a04219c87707b523 : SUCCESS in 1h 20m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87e106a2fc7b4e28bfc15013675d1e33 : FAILURE in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1689f62fb6234548937d1c0d13d16b7d : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3aa43aa130594c9a8d017368b4106b0c : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6af3fcd7152e440081846849b5565e88 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb9bbb114f95494e86901bf69c8bbf18 : SUCCESS in 4m 48s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/7af3c3b853b8462db3228d4e5669bc62 : FAILURE in 1h 27m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dda7ddf6849b046a509b8adad5df5dc5c3643c8e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-08-19 10:58:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a24a7f051d628da2b9c8df06faa01f8419889ba4","revisions":{"a2370425a212185058d7eeced1eb98f92ea0711c":{"kind":"REWORK","_number":1,"created":"2021-08-17 08:27:25.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/33/804833/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/33/804833/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/33/804833/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/33/804833/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/33/804833/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/33/804833/1"}}},"commit":{"parents":[{"commit":"ff8df6dc0e56e5c8305176140eb783bd9765b695","subject":"Drop lower-constraints job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ff8df6dc0e56e5c8305176140eb783bd9765b695"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-04-14 14:40:45.000000000","tz":-240},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-08-17 08:27:21.000000000","tz":330},"subject":"[DNM] Test lower constraints trigger","message":"[DNM] Test lower constraints trigger\n\nChange-Id: Ifaa4d39047a183c48a31ccb41e22534e1fc654a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a2370425a212185058d7eeced1eb98f92ea0711c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a2370425a212185058d7eeced1eb98f92ea0711c"}]},"branch":"refs/heads/stable/victoria"},"ff3504d007a49dbae2b3425a74515c33dfddad79":{"kind":"REWORK","_number":2,"created":"2021-08-17 08:43:49.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/33/804833/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/33/804833/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/33/804833/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/33/804833/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/33/804833/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/33/804833/2"}}},"commit":{"parents":[{"commit":"ff8df6dc0e56e5c8305176140eb783bd9765b695","subject":"Drop lower-constraints job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ff8df6dc0e56e5c8305176140eb783bd9765b695"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-04-14 14:40:45.000000000","tz":-240},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-08-17 08:43:14.000000000","tz":330},"subject":"[DNM] Test lower constraints trigger","message":"[DNM] Test lower constraints trigger\n\nChange-Id: Ifaa4d39047a183c48a31ccb41e22534e1fc654a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ff3504d007a49dbae2b3425a74515c33dfddad79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ff3504d007a49dbae2b3425a74515c33dfddad79"}]},"branch":"refs/heads/stable/victoria"},"a24a7f051d628da2b9c8df06faa01f8419889ba4":{"kind":"REWORK","_number":3,"created":"2021-08-17 08:47:16.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/33/804833/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/33/804833/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/33/804833/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/33/804833/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/33/804833/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/33/804833/3"}}},"commit":{"parents":[{"commit":"ff8df6dc0e56e5c8305176140eb783bd9765b695","subject":"Drop lower-constraints job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ff8df6dc0e56e5c8305176140eb783bd9765b695"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-04-14 14:40:45.000000000","tz":-240},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-08-17 08:46:57.000000000","tz":330},"subject":"[DNM] Test lower constraints trigger","message":"[DNM] Test lower constraints trigger\n\nChange-Id: Ifaa4d39047a183c48a31ccb41e22534e1fc654a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a24a7f051d628da2b9c8df06faa01f8419889ba4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a24a7f051d628da2b9c8df06faa01f8419889ba4"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
