)]}'
{"id":"openstack%2Fglance_store~815009","triplet_id":"openstack%2Fglance_store~master~I1f8c9184761313f9fc5fda2f257e52233e0196d1","project":"openstack/glance_store","branch":"master","topic":"bug/1620999","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-22 19:12:23.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-08-22 17:42:49.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-08-11 15:10:28.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I1f8c9184761313f9fc5fda2f257e52233e0196d1","subject":"Swift: Honor *_domain_name parameters","status":"MERGED","created":"2021-10-21 14:17:59.000000000","updated":"2022-08-22 19:13:30.000000000","submitted":"2022-08-22 19:12:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815009-bug/1620999","meta_rev_id":"7bb9d4b7e5f8586a7085884a05c4185e5a8752fd","_number":815009,"virtual_id_number":815009,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-22 19:12:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-08-18 16:27:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-08-22 17:42:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-08-22 17:42:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-21 15:57:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-11 14:51:13.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-08-18 16:27:43.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"3ba62d4afe82ba1f3f61cbff5b3ec74515c1df90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-21 14:17:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce7992c7d3061d9f2670275a8f155d5805f5827a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 15:57:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f6650be5da204a2db5b598a337b43e63 : SUCCESS in 1h 34m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88ea8e577c354447960d2be21e4dc71d : FAILURE in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddc6d61323e244e6af5dceb8bd2ffba2 : FAILURE in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7302aab625d6410aa3fd66b8f2dab938 : FAILURE in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f78a1bdf2c845a3815eab333cc11d32 : FAILURE in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e74bff606ac4f3f863d3097f919af4b : FAILURE in 3m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72f1c392bb9d400c9e03550cf1203a8b : SUCCESS in 5m 09s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/2592d791ad4b4a96a4e3ef74867ab18d : SUCCESS in 11m 15s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/d31e36b9075b4481a65aba8b9f8f5039 : FAILURE in 24m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8893e2accfad31de4f69a8000a13c5c56c52bc2e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-24 04:40:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39e34f42203b96d7c9e678fc14aec69f95d561c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-24 06:51:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a06ad962a104a369dfc6a1c9850939f : SUCCESS in 1h 54m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ba911b0c77a49ac8e7c14629daa2dc4 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c02564a591d74d65b77b8061bf9555c7 : FAILURE in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9b4afcd824b48f79f5ef42bcfd8b425 : FAILURE in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88f4f7cafb8945d98f2eed3e9af29c51 : FAILURE in 3m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/199defdcfcb048079edaa5115584143d : FAILURE in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f276492477c5417c818b668970a673bc : SUCCESS in 3m 38s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/c30d71443512470ea52bb9634e71981a : SUCCESS in 8m 59s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/b748f961e5894b73b9502e9fcd44222e : SUCCESS in 2h 05m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dbdb4ec46c7ad5eae7d299eaf01d88c98be4924c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-24 13:06:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"51499966088cebfd874107817c953f5705f5628c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-24 14:43:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb1d944399c24c478c4e731a813e84d8 : SUCCESS in 1h 20m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faadb89eeb7f4ba097a4c66c121024f1 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ea09546f8f540748933cf09aaeb087e : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d333d8eee594e449599035336c1dabc : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec278af2739749e69d0695fbac238b4c : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbe1add887c344f89d6665be6e2d0aab : FAILURE in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45e01e5e50a94ec0a26b1d5afa2a6e46 : SUCCESS in 5m 27s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/8394d3ec5b684a869e8a30b9ed55c789 : SUCCESS in 9m 35s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/cdaec9c24b7e47fb97585801c76521e6 : SUCCESS in 1h 35m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"12ca0c4a9c57407d8688b71f785bd599a4a4069f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-15 10:56:16.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0b90592a3e0f80a32027187724d577418eabd01a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 12:45:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eea48da7baa24ce9a2fbabce71f71bb4 : SUCCESS in 1h 44m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ece0db50b11945a4aa74c9cd02912c49 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d722cb66e60b4226a7b8b95e13737a77 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f28bb649d4684dda9493c70d1c282cbb : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/645313226fa54c7cac98f6c2cb5a4646 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0a7dea438f84c7185271234b0a32cdd : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fecf6531b7864f15886251f30d2b2c9e : SUCCESS in 4m 28s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/ab21419ac2de4ae9ae2f27719158594d : SUCCESS in 9m 09s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/cd0679c8a5fe4e3ea12fe6cd1d4ad2c4 : SUCCESS in 1h 35m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c3f9c10c2fb9c3f4f6f64f1d2ecb054c827dd65b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-08-05 14:16:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5cef943485c473d3ec61001b3a1e4a35f3258568","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 16:27:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd5473d18e6d424291bc9dbf4c2c015f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/38b55d66acd4497ca5ddac44aabaf865 : SUCCESS in 1h 46m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b2b5b49841b4c5cbd8095d3ff26063c : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa4d3d99edb04d9eb42cd9faad1eb533 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98c19e48cdf64553b9b0d4ef8e5e9a5e : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bbd62f151d0d4e15a3bde663cd232c8a : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d649d814cf8c4367a0041ce92d57a91b : SUCCESS in 4m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/234b802a4c7c48d18b9b5baa4a504ebc : SUCCESS in 3m 55s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/2d3e64d19b134a6fa5c0fee211f824b8 : SUCCESS in 10m 17s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/08b6757f510d4bfe859aa23be80ff56c : TIMED_OUT in 2h 09m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"93d10407eb360eecbd2335a41d59e08f3157eead","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-08-11 14:51:13.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3e9b083a0d7ba48b2c5769bfcf0f8b2a1870d8af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-11 15:10:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c3a416dd38dba47584c489f3e43d996da4c99248","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-11 15:10:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a7eaf2235b70b89509cc4dc094f14873920d54cf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-11 15:14:43.000000000","message":"Patch Set 5: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fe045426133a0b66e83126b066ed943fa063dff2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-11 16:18:29.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c7710239cba0bb8ea53ce167c67f2071d89f0c40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 18:07:48.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/06c3c140b6c24e9aa84c3f19783436b9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05f27dd0f30840428a6df681fea834fe : SUCCESS in 1h 39m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82e7cf621d7b4300abd08a456809bb40 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b1d3754d2cd4b84a08688debdce76eb : FAILURE in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0dcc66f6aac4680b6613003618b1c7b : FAILURE in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c83c7e6e72da483fafad87706b2d54d6 : FAILURE in 3m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ff7cc302ffc44e182e2fe318d4cdcf3 : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4561abc76234e03aee16db56ce3a2ce : SUCCESS in 4m 17s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/0164303664a446afb72c2a3791a414c7 : SUCCESS in 10m 07s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/69970bc6525c4c62a546861012de3a39 : SUCCESS in 1h 47m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"46289e19a9b5a47022f79f46bd48c5af5b85d800","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-18 02:17:08.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1f8546c17945743438f05c9d6118a3a9380b9597","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-18 02:21:05.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"587d3330653e00d44c5285c85845d5bcae1de2fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 04:04:26.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/276d107c67af422da2e7473974c9fdde\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4aa5506cfe649929d29c0f6429ddb5f : SUCCESS in 1h 18m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f602a5df25914b748bc263203e2e46e4 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/069302657cac4d60891c386bfbb8fc8b : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/031628d1119a4753bc9980c673fe42d1 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26c4bc16eac747a89614185970eff5af : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c03bce1569db44449924a2ba01517dca : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc41cf033d1141b0b843a14b9c04ae75 : SUCCESS in 4m 12s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/1131b0bc991b4e01bbb7430a79d31890 : SUCCESS in 11m 05s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/6008bb747e5f4db18bd103df9109fcc0 : SUCCESS in 1h 37m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"87b7e94a53060f7090d61f54e7d85da09e621220","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-18 16:27:43.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c89c7313bfa9e394e055cc07ec8ddd45a62b0db8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-08-22 17:42:49.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"42149c5170018de4a9857b90de34219de2a7e1c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 17:43:25.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"b789c017c98de263bfbc21cecae402075005db3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 19:12:23.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95f93e98cccd4f2b9cb5d77de4e410cb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c407c3e5d7e4fde89b4289523c57ad7 : SUCCESS in 1h 28m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89287c9c6f3148df9c6a7f2bdc641554 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb4c1742fd89420996d593fef1ea0b31 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3366b803fafd4a5898a5d10bca8c9ada : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be70433ef017410fbe8deedfc9d83482 : SUCCESS in 3m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98e21aa566ba4646b5374e17e809fe32 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":8},{"id":"2d9fbdf511046be8e0cca3bc4002e1bd8326d293","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 19:12:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"7bb9d4b7e5f8586a7085884a05c4185e5a8752fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 19:13:30.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc02e5b5039448c6ba809b58e4bcb076\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27cce287265642f29f288b11af0f6442 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/451bb3f748064bbeb14895858a2b906f : SUCCESS in 49s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"6738d0b156fc1caae0604eb7bd8f05a6ed1c2d68","revisions":{"5173d3d4d5f0cbb49f70d38cbe863eb91cb5043d":{"kind":"REWORK","_number":1,"created":"2021-10-21 14:17:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/815009/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/09/815009/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/09/815009/1"}}},"commit":{"parents":[{"commit":"68200a88ff03921f47c2fad872a7f7e451fd1638","subject":"Add Python3 yoga unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/68200a88ff03921f47c2fad872a7f7e451fd1638"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-21 13:57:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-21 13:57:52.000000000","tz":540},"subject":"Swift: Honor *_domain_name parameters","message":"Swift: Honor *_domain_name parameters\n\nThe *_domain_id parmaeters should not have any default. Otherwise\nkeystoneauth ignores the *_domain_name parameters and it requires\nonly *_domain_id parameters are used.\n\nCloses-Bug: #1620999\nChange-Id: I1f8c9184761313f9fc5fda2f257e52233e0196d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/5173d3d4d5f0cbb49f70d38cbe863eb91cb5043d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/5173d3d4d5f0cbb49f70d38cbe863eb91cb5043d"}]},"branch":"refs/heads/master"},"f0fd7992fcabdf09fc850557626c7509b599975d":{"kind":"REWORK","_number":2,"created":"2021-10-24 13:06:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/815009/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/09/815009/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/09/815009/2"}}},"commit":{"parents":[{"commit":"68200a88ff03921f47c2fad872a7f7e451fd1638","subject":"Add Python3 yoga unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/68200a88ff03921f47c2fad872a7f7e451fd1638"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-21 13:57:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-24 13:05:48.000000000","tz":540},"subject":"Swift: Honor *_domain_name parameters","message":"Swift: Honor *_domain_name parameters\n\nThe *_domain_id parmaeters should not have any default. Otherwise\nkeystoneauth ignores the *_domain_name parameters and it requires\nonly *_domain_id parameters are used.\n\nCloses-Bug: #1620999\nChange-Id: I1f8c9184761313f9fc5fda2f257e52233e0196d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/f0fd7992fcabdf09fc850557626c7509b599975d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/f0fd7992fcabdf09fc850557626c7509b599975d"}]},"branch":"refs/heads/master"},"091d992b2aa10fdb17f771d4b1a282cd19d8f507":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-11-15 10:56:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/815009/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/09/815009/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/09/815009/3"}}},"commit":{"parents":[{"commit":"e4930e5dc8b7e3102a4f32ffa6a18af8b4cb3a3a","subject":"Fix documentation build with Sphinx\u003e\u003d4.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/e4930e5dc8b7e3102a4f32ffa6a18af8b4cb3a3a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-21 13:57:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-15 10:56:16.000000000","tz":0},"subject":"Swift: Honor *_domain_name parameters","message":"Swift: Honor *_domain_name parameters\n\nThe *_domain_id parmaeters should not have any default. Otherwise\nkeystoneauth ignores the *_domain_name parameters and it requires\nonly *_domain_id parameters are used.\n\nCloses-Bug: #1620999\nChange-Id: I1f8c9184761313f9fc5fda2f257e52233e0196d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/091d992b2aa10fdb17f771d4b1a282cd19d8f507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/091d992b2aa10fdb17f771d4b1a282cd19d8f507"}]},"branch":"refs/heads/master","description":"Rebase"},"e7294f6af178ddc1a09ba2d74dbb8871a6c4b35e":{"kind":"REWORK","_number":4,"created":"2022-08-05 14:16:30.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/09/815009/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/09/815009/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/09/815009/4"}}},"commit":{"parents":[{"commit":"6851cab51a3d5dbf018f755efee8e0e640ac12ab","subject":"Add debug logs to cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6851cab51a3d5dbf018f755efee8e0e640ac12ab"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-21 13:57:52.000000000","tz":540},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-08-05 14:15:54.000000000","tz":120},"subject":"Swift: Honor *_domain_name parameters","message":"Swift: Honor *_domain_name parameters\n\nThe *_domain_id parmaeters should not have any default. Otherwise\nkeystoneauth ignores the *_domain_name parameters and it requires\nonly *_domain_id parameters are used.\n\nCloses-Bug: #1620999\nChange-Id: I1f8c9184761313f9fc5fda2f257e52233e0196d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/e7294f6af178ddc1a09ba2d74dbb8871a6c4b35e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/e7294f6af178ddc1a09ba2d74dbb8871a6c4b35e"}]},"branch":"refs/heads/master"},"d216d462b93f4fdbe2f86d2ff4ff0eeb8296c939":{"kind":"REWORK","_number":5,"created":"2022-08-11 15:10:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/815009/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/09/815009/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/09/815009/5"}}},"commit":{"parents":[{"commit":"6851cab51a3d5dbf018f755efee8e0e640ac12ab","subject":"Add debug logs to cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6851cab51a3d5dbf018f755efee8e0e640ac12ab"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-21 13:57:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-11 15:09:35.000000000","tz":540},"subject":"Swift: Honor *_domain_name parameters","message":"Swift: Honor *_domain_name parameters\n\nThe *_domain_id parmaeters should not have any default. Otherwise\nkeystoneauth ignores the *_domain_name parameters and it requires\nonly *_domain_id parameters are used.\n\nCloses-Bug: #1620999\nChange-Id: I1f8c9184761313f9fc5fda2f257e52233e0196d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d216d462b93f4fdbe2f86d2ff4ff0eeb8296c939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d216d462b93f4fdbe2f86d2ff4ff0eeb8296c939"}]},"branch":"refs/heads/master"},"44bfa843f703108ba34299642e710ee8bd0dafc1":{"kind":"REWORK","_number":6,"created":"2022-08-11 16:18:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/815009/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/09/815009/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/09/815009/6"}}},"commit":{"parents":[{"commit":"6851cab51a3d5dbf018f755efee8e0e640ac12ab","subject":"Add debug logs to cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6851cab51a3d5dbf018f755efee8e0e640ac12ab"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-21 13:57:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-11 16:18:12.000000000","tz":540},"subject":"Swift: Honor *_domain_name parameters","message":"Swift: Honor *_domain_name parameters\n\nThe *_domain_id parmaeters should not have any default. Otherwise\nkeystoneauth ignores the *_domain_name parameters and it requires\nonly *_domain_id parameters are used.\n\nCloses-Bug: #1620999\nChange-Id: I1f8c9184761313f9fc5fda2f257e52233e0196d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/44bfa843f703108ba34299642e710ee8bd0dafc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/44bfa843f703108ba34299642e710ee8bd0dafc1"}]},"branch":"refs/heads/master"},"81bc1a3d929a75c81f5abfeeff4ec2c69b1f4a2a":{"kind":"REWORK","_number":7,"created":"2022-08-18 02:17:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/815009/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/09/815009/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/09/815009/7"}}},"commit":{"parents":[{"commit":"6851cab51a3d5dbf018f755efee8e0e640ac12ab","subject":"Add debug logs to cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6851cab51a3d5dbf018f755efee8e0e640ac12ab"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-21 13:57:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-18 02:16:57.000000000","tz":540},"subject":"Swift: Honor *_domain_name parameters","message":"Swift: Honor *_domain_name parameters\n\nThe *_domain_id parmaeters should not have any default. Otherwise\nkeystoneauth ignores the *_domain_name parameters and it requires\nonly *_domain_id parameters are used.\n\nCloses-Bug: #1620999\nChange-Id: I1f8c9184761313f9fc5fda2f257e52233e0196d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/81bc1a3d929a75c81f5abfeeff4ec2c69b1f4a2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/81bc1a3d929a75c81f5abfeeff4ec2c69b1f4a2a"}]},"branch":"refs/heads/master"},"6738d0b156fc1caae0604eb7bd8f05a6ed1c2d68":{"kind":"REWORK","_number":8,"created":"2022-08-18 02:21:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/815009/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/09/815009/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/815009/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/09/815009/8"}}},"commit":{"parents":[{"commit":"6851cab51a3d5dbf018f755efee8e0e640ac12ab","subject":"Add debug logs to cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6851cab51a3d5dbf018f755efee8e0e640ac12ab"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-21 13:57:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-18 02:20:52.000000000","tz":540},"subject":"Swift: Honor *_domain_name parameters","message":"Swift: Honor *_domain_name parameters\n\nThe *_domain_id parmaeters should not have any default. Otherwise\nkeystoneauth ignores the *_domain_name parameters and it requires\nonly *_domain_id parameters are used.\n\nCloses-Bug: #1620999\nChange-Id: I1f8c9184761313f9fc5fda2f257e52233e0196d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6738d0b156fc1caae0604eb7bd8f05a6ed1c2d68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6738d0b156fc1caae0604eb7bd8f05a6ed1c2d68"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
