)]}'
{"id":"openstack%2Fglance_store~829771","triplet_id":"openstack%2Fglance_store~stable%2Fwallaby~I206580f6be615ebc5e15b546b9c23728d4116a5d","project":"openstack/glance_store","branch":"stable/wallaby","topic":"826290-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-02-18 16:24:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I206580f6be615ebc5e15b546b9c23728d4116a5d","subject":"Cinder store: Wait for device resize","status":"MERGED","created":"2022-02-17 18:31:53.000000000","updated":"2022-02-18 16:26:03.000000000","submitted":"2022-02-18 16:24:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829771-826290-stable/xena-stable/wallaby","cherry_pick_of_change":829705,"cherry_pick_of_patch_set":1,"meta_rev_id":"2b80a4b1ad8e8bd86ed73cb3a992fed8a55f0b13","_number":829771,"virtual_id_number":829771,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-18 16:24:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-02-18 14:57:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2022-02-17 18:40:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-02-18 14:57:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-17 18:40:12.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-02-17 20:02:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-18 14:57:04.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"e0a9a37efd7ad27889b40ae1284502c8887a2ee9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-02-17 18:31:53.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"f6c9af19bd0d22e09e3a4963f11f45a5ab53770d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-17 18:40:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8fe5510d67c867f5ff6904b6ec9c4aa6088ba29c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 20:02:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f22852074f164695906a6e7aed75363b : SUCCESS in 1h 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a47a667128714eb7b3ae4e2fb891e4d8 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f3588082e644054915b50106c40a64a : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83291e4cd1774aca929ba1101f26637c : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c2fae7135b8461696e66a08649a32c6 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca3a004391bb44088d7550c9ae8f1a24 : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7af8bd8147e42f3b74b8e49d027898e : SUCCESS in 3m 53s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/2f62e56f328d4fa6b6d4451e9e3787b8 : SUCCESS in 8m 46s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/dea4feaa4b6a490187af22e9ecd63767 : FAILURE in 1h 28m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9d0348fd8bbc695690acee4ab4446de9dd9e0f8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-18 14:57:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae84d7c7156a433a5165d5c95d9e12c234ccd7da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 14:57:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ea733d1509ea7fd4eb6c320a449a402ebdf3bc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 16:24:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6eacabe5a13e494a90c2a0af0b663cb2 : SUCCESS in 1h 26m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/189179980e804acdb8d92c7c39e0073f : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76f1fc659fe640ea9d4e5074d2c29329 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7234aa38e7ea4ed6b426c66563f598cc : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caac0100041f4578b285101bffe03bc6 : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79ab27b6612a40d4aa469747e3a0854a : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"9efc64e66e18b08d255dd6d4f501936e4a6c94f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 16:24:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2b80a4b1ad8e8bd86ed73cb3a992fed8a55f0b13","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 16:26:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0239818829234602bf2630ec244b32f1 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a14994d0e13a4aba9efc72321723f7f3 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f1cee6fa90268b17c7b6f842854a3b27e44994c","revisions":{"5f1cee6fa90268b17c7b6f842854a3b27e44994c":{"kind":"REWORK","_number":1,"created":"2022-02-17 18:31:53.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/71/829771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/71/829771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/71/829771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/71/829771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/71/829771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/71/829771/1"}}},"commit":{"parents":[{"commit":"2bc17c0a9e35882d047c20acd2da603a052ce099","subject":"Pass valid IP address to os-brick","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2bc17c0a9e35882d047c20acd2da603a052ce099"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-01-25 13:54:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-02-17 18:31:53.000000000","tz":0},"subject":"Cinder store: Wait for device resize","message":"Cinder store: Wait for device resize\n\nWhen we have an image with size \u003e 1 GB, we follow the following steps\nto accomodate the image:\n1) Detach the volume\n2) extend the volume\n3) Attach the volume\n4) Open the volume device as a file and resume writing the image\n\nSometimes due to several reasons (mostly network related), the\nsize of the device file could mismatch with the actual volume size\n(or the backend LUN size). This can happen if the extend was performed\n(i.e. the control path) but it takes the time to reflect that into\nthe mapped device (i.e. the data path). This mismatch can cause the\nissue \"IOError: [Errno 28] No space left on device\".\nTo avoid this scenario, we check if the device size is less than the\nvolume size, we wait for the extended LUN to show up in mapped device\nand then continue the image writing operation.\n\nCloses-Bug: #1959913\n\nChange-Id: I206580f6be615ebc5e15b546b9c23728d4116a5d\n(cherry picked from commit f3433ed1a5176bff85b3fe04dba2d4c76618a299)\n(cherry picked from commit ea5139be9a76ca8c2e60d12d8dd776bff39c4579)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/5f1cee6fa90268b17c7b6f842854a3b27e44994c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/5f1cee6fa90268b17c7b6f842854a3b27e44994c"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
