)]}'
{"id":"openstack%2Fglance_store~833230","triplet_id":"openstack%2Fglance_store~master~I4deb193b44394ae9f99f57e31fb81fe04394e247","project":"openstack/glance_store","branch":"master","topic":"improve-coverage-cinder-store","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-03-16 03:23:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4deb193b44394ae9f99f57e31fb81fe04394e247","subject":"Remove six usage","status":"MERGED","created":"2022-03-11 09:08:27.000000000","updated":"2022-03-16 03:25:24.000000000","submitted":"2022-03-16 03:23:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833230-improve-coverage-cinder-store","meta_rev_id":"09a2924a38ed20ab23b2248b2c634d231c41f87e","_number":833230,"virtual_id_number":833230,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-16 03:23:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-03-16 03:25:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-03-16 01:41:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-15 02:33:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2022-03-16 01:41:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-11 11:04:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-14 14:17:12.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-03-15 02:33:53.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-03-16 01:41:17.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"3f93a20f273b61529a7f0b1c4366534194824bf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-03-11 09:08:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69c5ea680b524719f144bc227e8f3e9dc3953be9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 11:04:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0155528e95a84e74880fa7187b331303 : SUCCESS in 1h 45m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c70ed9df55794be88fa64bb5cd2a5f21 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bdf6f8716f24f56a5d04c669ccf9401 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57b294bf1eab489ab29d12fbd40f0719 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acb65b85bf664a8497d2281e20e55286 : SUCCESS in 5m 24s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/3c154de2ba1a4909a238e2ebfe997926 : SUCCESS in 10m 49s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/d67eee6e190642e697585470694f87eb : SUCCESS in 1h 47m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed8b92a28fff3d5a6e593cb2978305113cbc30b0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-03-14 09:14:38.000000000","message":"Topic 832922 removed","accounts_in_message":[],"_revision_number":1},{"id":"b2a4c660da9900d8c469372bcfa5ef1081bd2fe3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-03-14 09:14:45.000000000","message":"Topic set to improve-coverage-cinder-store","accounts_in_message":[],"_revision_number":1},{"id":"7fe0d71a6de956c3c8070be9d8f0f14eb5af8943","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-03-14 14:17:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"63b9c5e3e2f7e1b2e32bdf88094d39ff03634c2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-03-14 16:56:43.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"5d8e8072a35b0d88c7c8b68f8de48c8fa1c00b12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 18:30:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e0a68f024e84258a4f85590ee177d4d : SUCCESS in 1h 27m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6564e535f1904c5d95a4137e744bf2a3 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ea7145f223d496c9ce342c987376350 : SUCCESS in 8m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56f4d763412540d6803640e1bc65e4d5 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff43dde1e48c4b71aabceef523a71fbf : SUCCESS in 6m 12s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/b541b327710848d09db90d2bb79af890 : SUCCESS in 12m 22s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/07debb389a0b4ff2a280eaea128db0ca : FAILURE in 1h 01m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bbb87959cd9285503867955ba95cfc9d833e12d8","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-03-15 02:33:53.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0470bdb8a595d0f4323c9d467ae8bc4a89d7ff9c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-03-16 01:41:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"25fdb2bf854f238106f10acdb09b8ce07b3a4b0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 01:42:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3d063ded389b50366addf9c5a653bd67a54e0fd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 03:23:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b1ddfb336d44777ba6bd41eadcf97d2 : SUCCESS in 1h 35m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14abc01f395c40858af7d8bf2cbcb6fd : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/164c9e35b15b4dc09f063c6f8f95b7ad : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a4e403c1d4048cea182baedd584801f : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0f291ac70614d3eb02cb9f50e82ff88 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":2},{"id":"5ba40605c021c9a6049923f0ce50f4efc8d89352","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 03:23:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f30f68ef78683b6d8ac575af2db0b380c57b9a5e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 03:25:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21f0013784ef4736b9d9ef85ab9058cc : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"95b49304cde62ffbddb4e47b782320cbf7e65ca0","revisions":{"1d4fc35cf70b0103586be74039fc6b57ae126f74":{"kind":"REWORK","_number":1,"created":"2022-03-11 09:08:27.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/30/833230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/30/833230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/30/833230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/30/833230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/30/833230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/30/833230/1"}}},"commit":{"parents":[{"commit":"fb10a8c197ad05e1f00e23d0b2040731f3fcea15","subject":"Refactor cinder store tests[2/2]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/fb10a8c197ad05e1f00e23d0b2040731f3fcea15"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-03-11 09:04:52.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-03-11 09:04:52.000000000","tz":330},"subject":"Remove six usage","message":"Remove six usage\n\nsix has been used to maintain compatibility between python2 and\npython3 code. Since current openstack development branches doesn\u0027t\nsupport python2, we don\u0027t need the six library and can use python3\ncode directly.\nThis patch removes six usage from cinder store tests.\n\nChange-Id: I4deb193b44394ae9f99f57e31fb81fe04394e247\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/1d4fc35cf70b0103586be74039fc6b57ae126f74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/1d4fc35cf70b0103586be74039fc6b57ae126f74"}]},"branch":"refs/heads/master"},"95b49304cde62ffbddb4e47b782320cbf7e65ca0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-14 16:56:43.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/30/833230/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/30/833230/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/30/833230/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/30/833230/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/30/833230/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/30/833230/2"}}},"commit":{"parents":[{"commit":"fba6d0dd836a6d5741bceddfb4fa37c4af7daff6","subject":"Refactor cinder store tests[2/2]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/fba6d0dd836a6d5741bceddfb4fa37c4af7daff6"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-03-11 09:04:52.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-03-14 16:40:52.000000000","tz":330},"subject":"Remove six usage","message":"Remove six usage\n\nsix has been used to maintain compatibility between python2 and\npython3 code. Since current openstack development branches doesn\u0027t\nsupport python2, we don\u0027t need the six library and can use python3\ncode directly.\nThis patch removes six usage from cinder store tests.\n\nChange-Id: I4deb193b44394ae9f99f57e31fb81fe04394e247\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/95b49304cde62ffbddb4e47b782320cbf7e65ca0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/95b49304cde62ffbddb4e47b782320cbf7e65ca0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
