)]}'
{"id":"openstack%2Fglance_store~833751","triplet_id":"openstack%2Fglance_store~master~I18b1e5e8fd818824a2dda2ad14d9456190fe9ff4","project":"openstack/glance_store","branch":"master","topic":"improve-coverage-cinder-store","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-03-24 05:58:15.000000000","reason":"Abhishek Kekane replied on the change"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-03-24 07:23:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I18b1e5e8fd818824a2dda2ad14d9456190fe9ff4","subject":"Add coverage for get_cinderclient and _check_context","status":"MERGED","created":"2022-03-15 09:14:57.000000000","updated":"2022-03-24 07:24:45.000000000","submitted":"2022-03-24 07:23:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833751-improve-coverage-cinder-store","meta_rev_id":"0b2aa6caa3dcda291e3bb84e1babe95b5da6ed13","_number":833751,"virtual_id_number":833751,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-24 07:23:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-03-24 05:58:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-03-23 20:50:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-03-24 05:58:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-15 10:54:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-16 20:12:41.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-03-22 07:54:27.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2022-03-24 05:58:15.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"1b443cdc070ddeec74c6a69c1ca76b90564644e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-03-15 09:14:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9862507e7bef466fd57ffb54bb28e0e93f336bdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 10:54:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68c45cb9798c448885807ad27948af8a : SUCCESS in 1h 37m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b281610919148509603af67aa86727b : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a5bc194c60f4192a3c26c840027d26c : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81cf6adf11ae40f8af431259b8b83da0 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efd31aeb525442e590326d794adfbfac : SUCCESS in 5m 20s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/ab7d464e49c64240843a3540e5e435bf : SUCCESS in 11m 29s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/28337a03ffc54d87850b2164e725a5ef : FAILURE in 59m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1200953b29c5d7acf7d39d8f76ef65f23598da32","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-03-16 20:12:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c3d3a7b429ef104140494a4a6c6a8991de151c48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-03-17 10:39:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"5db52f597820560e3d66f2dbb15fe0c37c870062","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 12:18:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/32f902c0b0054a309339755cada380df : SUCCESS in 1h 25m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f08b7d1e37549e5958628556fa523b7 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aff3830073d0445cbd9638bc8cd07aa5 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65c668d86a164e6ab88e8ae539ddc7d8 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53121e863b0f445ba95602db2ae27e12 : SUCCESS in 5m 42s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/807ae24a9fb8470a839cc7568e74ea51 : SUCCESS in 10m 23s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/5cf0b04d518d4ebfad8086106ff714d9 : SUCCESS in 1h 32m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9d15e6f7f3283dfb28c61e71014ee180c1532e3e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-03-22 07:54:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"209b741bfcfa0410d601283bf09d1976a6509dce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-03-22 09:08:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"84a6163ffd309f63bc5656d01226ed9fd38045df","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-03-22 09:08:26.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1eea181d50ba94f0b93d6de27ddd4cb6e7b98364","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 10:54:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea5ea644b24d4301b949f380a68bc506 : SUCCESS in 1h 42m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/659e4f860c5041dc9b4f447c1bab6551 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f7f15443e9d424a9599199252935af8 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c82b8bfcce84ac6a36ccc568d758d72 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/071b7fe175e345ceb30586e678e19a03 : SUCCESS in 5m 06s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/1e93c75e45d34875bbd620c78826bb64 : SUCCESS in 10m 36s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/b93e11301beb401d92a5df4946de93df : SUCCESS in 1h 40m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"24364a586fe1469794609b9ee79643a1bddb7242","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-03-23 20:50:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"97fbcfb9a191614d94e2271d0cdd7b6c42c00d41","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-03-24 05:58:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"218be71cc2a2c2859424b7ea515914f094450391","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 05:58:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"28fc6f18deeccff74c1b9e45a68bb39d5981fc4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 07:23:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28f6b4dfab4e4d71bbb14376bf842474 : SUCCESS in 1h 23m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e61c53b01fd4ab2be6b2b205a19aeea : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fa8e8d9f0a74ee4a3d565cf23e088d7 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ba76d5cb39f45098d15da76647c4ceb : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1f684256330448f83a4de3cab41af51 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":3},{"id":"0494748495b8be6820ceec01a489b7f7861ec8bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 07:23:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0b2aa6caa3dcda291e3bb84e1babe95b5da6ed13","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 07:24:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bff5ef66f654ce7ae6666b12e5564a3 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"77919e15d281e908de7b687e2af927b59d3dc8a8","revisions":{"52abd39b67b1cff6781be04c32769c2b0ef572a9":{"kind":"REWORK","_number":1,"created":"2022-03-15 09:14:57.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/51/833751/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/51/833751/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/51/833751/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/51/833751/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/51/833751/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/51/833751/1"}}},"commit":{"parents":[{"commit":"64aae33ec60c83a96ac065c236bb940996ce3686","subject":"Add coverage for StoreLocation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/64aae33ec60c83a96ac065c236bb940996ce3686"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-03-15 08:51:13.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-03-15 08:51:13.000000000","tz":330},"subject":"Add coverage for get_cinderclient and _check_context","message":"Add coverage for get_cinderclient and _check_context\n\nThis patch adds coverage for some parts in get_cinderclient\nmethod like ``cinder_endpoint_template`` option and keystone\nexception case when catalog info for cinder could not be found.\nAlso it adds tests for complete coverage of _check_context\nmethod.\n\nChange-Id: I18b1e5e8fd818824a2dda2ad14d9456190fe9ff4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/52abd39b67b1cff6781be04c32769c2b0ef572a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/52abd39b67b1cff6781be04c32769c2b0ef572a9"}]},"branch":"refs/heads/master"},"9d08e162ffb9f84346ab8fc2928c9c90aab27d48":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-17 10:39:50.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/51/833751/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/51/833751/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/51/833751/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/51/833751/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/51/833751/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/51/833751/2"}}},"commit":{"parents":[{"commit":"3666d2d2e4be4524c47ba2531e39984e260c8ce8","subject":"Add coverage for StoreLocation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3666d2d2e4be4524c47ba2531e39984e260c8ce8"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-03-15 08:51:13.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-03-17 08:59:51.000000000","tz":330},"subject":"Add coverage for get_cinderclient and _check_context","message":"Add coverage for get_cinderclient and _check_context\n\nThis patch adds coverage for some parts in get_cinderclient\nmethod like ``cinder_endpoint_template`` option and keystone\nexception case when catalog info for cinder could not be found.\nAlso it adds tests for complete coverage of _check_context\nmethod.\n\nChange-Id: I18b1e5e8fd818824a2dda2ad14d9456190fe9ff4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/9d08e162ffb9f84346ab8fc2928c9c90aab27d48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/9d08e162ffb9f84346ab8fc2928c9c90aab27d48"}]},"branch":"refs/heads/master"},"77919e15d281e908de7b687e2af927b59d3dc8a8":{"kind":"REWORK","_number":3,"created":"2022-03-22 09:08:05.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/51/833751/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/51/833751/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/51/833751/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/51/833751/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/51/833751/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/51/833751/3"}}},"commit":{"parents":[{"commit":"3666d2d2e4be4524c47ba2531e39984e260c8ce8","subject":"Add coverage for StoreLocation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3666d2d2e4be4524c47ba2531e39984e260c8ce8"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-03-15 08:51:13.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-03-22 08:48:18.000000000","tz":330},"subject":"Add coverage for get_cinderclient and _check_context","message":"Add coverage for get_cinderclient and _check_context\n\nThis patch adds coverage for some parts in get_cinderclient\nmethod like ``cinder_endpoint_template`` option and keystone\nexception case when catalog info for cinder could not be found.\nAlso it adds tests for complete coverage of _check_context\nmethod.\n\nChange-Id: I18b1e5e8fd818824a2dda2ad14d9456190fe9ff4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/77919e15d281e908de7b687e2af927b59d3dc8a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/77919e15d281e908de7b687e2af927b59d3dc8a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
