)]}'
{"id":"openstack%2Fglance_store~833982","triplet_id":"openstack%2Fglance_store~master~If63a6c810b5aab992e54857bc81f5052c2c593c4","project":"openstack/glance_store","branch":"master","topic":"improve-coverage-cinder-store","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-03-24 05:58:52.000000000","reason":"Abhishek Kekane replied on the change"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-05-04 19:11:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If63a6c810b5aab992e54857bc81f5052c2c593c4","subject":"Add exception coverage for get, get_size, delete","status":"MERGED","created":"2022-03-16 10:16:47.000000000","updated":"2022-05-04 19:13:06.000000000","submitted":"2022-05-04 19:11:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833982-improve-coverage-cinder-store","meta_rev_id":"23434f1bacad338674e153a0dc0a4ed5674df0a7","_number":833982,"virtual_id_number":833982,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-04 19:11:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-03-24 05:58:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-05-04 17:39:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2022-05-04 17:39:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-16 12:01:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-22 08:07:57.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2022-03-24 05:58:52.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-05-04 17:39:11.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"ae64d159a40a76623037396c84808ee8c244b312","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-03-16 10:16:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79b758bda86e8914cc88e42368b48e53a0e8167e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 12:01:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/340dcc5e1a48459db8566565ac887e1c : SUCCESS in 1h 27m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83e01bd865fb41c8a7b22527b37b94c2 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29acce3e0aa4433ba37ecebaf5ee2576 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba5bb1abc8cd4fd6b31d129fd2ba42c3 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dce19d9c079b4866b711093f703fba4a : SUCCESS in 4m 54s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/1b967c9e0d2c4016a77d2e0b6c9cc4e4 : SUCCESS in 9m 39s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/c1998bc428f2475c803c7b568693b116 : SUCCESS in 1h 38m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ea2e8b1976b1a25b805b09beba2fd027cb0a431","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-03-17 10:39:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"03db3ee1e31813c96c21569a61816f74031ca208","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 12:23:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8298feab5ce94c1abaf517b877a2c79c : SUCCESS in 1h 23m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d46b4cd3adff41f1a3f6459a579ad686 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41bd7a4315924aaeb22e138787baef6c : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c88288dd9b04ff394c5ec58650cb76f : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f1820e52cc84901b1e29894785b6356 : SUCCESS in 6m 09s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/9823f30304df423f9d48da30decb6d11 : SUCCESS in 10m 53s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/7bdc86af00784c6db54529b33e8feaf0 : FAILURE in 1h 37m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"37acfc11e2cfa09738d252fd6951a191cc4e6e8e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-03-22 08:07:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"74f90945b49ea06d31827b78d94257384f3c72f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-03-22 09:08:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"45a2858afb8b45ad7f9de90c70341b17bd7d88af","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-03-22 09:08:33.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"54e9e0755a34550c7c103d8c2870f9825f59d378","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 11:08:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f32c9252cb184d58bc80377d5a4a25c0 : SUCCESS in 1h 31m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b10d5850e8bf4803816f922031dc9762 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5813f72e2d3412b8680c09d3959b6e6 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82fb4b5b42dc4459bff93736593744e9 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6094edc4b354afd853ee8dc5742f3a9 : SUCCESS in 5m 03s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/f21d8c6b2d3a4ca2ac01fb5ee3398079 : SUCCESS in 11m 29s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/48b624492cc14a4d8bcc56cc7d2ea3db : SUCCESS in 1h 52m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bf45d110b19dba79436e02908f8b7ca01a3bee49","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-03-24 05:58:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"afd7ea329943b6a6f80e2b901d6dd3d230892009","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-05-04 17:39:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9230ab0cee74832bb5b7b6d01e38b550fbb66966","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 17:39:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0b8d40cd5cbfd219e05ba8a19906a719ff441d50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 19:11:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff3f5801fc7f4bccac7c509e8c72ab4a : SUCCESS in 1h 27m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30f65830019048b784088f53a53f825c : SUCCESS in 2m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06d6aefcbf144def93383465638c0006 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26aa703682ed47bdaf74b79585474222 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94108b679d5640cdb8d4d95e218fdb21 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":3},{"id":"eb49ace9c19acf00de4a37abce8b78bda89422d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 19:11:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"23434f1bacad338674e153a0dc0a4ed5674df0a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 19:13:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/055fcf329312417bb8420e0b30ccb157 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4bd0304efbdfb98e2742741af42b483708f48a41","revisions":{"6f018fd4ec0c6a6f4719b0d547af0cd53cbdffd4":{"kind":"REWORK","_number":1,"created":"2022-03-16 10:16:47.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/82/833982/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/82/833982/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/833982/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/833982/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/833982/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/82/833982/1"}}},"commit":{"parents":[{"commit":"f27ebb4d49dc378b47c3ac682ad0c160a7247222","subject":"Add coverage for helper methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/f27ebb4d49dc378b47c3ac682ad0c160a7247222"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-03-16 10:10:04.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-03-16 10:10:04.000000000","tz":330},"subject":"Add exception coverage for get, get_size, delete","message":"Add exception coverage for get, get_size, delete\n\nThis patch adds coverage for the various exceptions handled and\nraised in get, get_size and delete methods.\nIt also corrects the behavior of _test_cinder_get_size method where\nclient.volumes returned dictionary instead of a MagicMock and it\nworked due to the existing method \"get\" in dictionary which has\nsame name as \"get\" method in cinderclient. The dictionary object\nis replaced with the appropriate MagicMock object in this patch.\n\nChange-Id: If63a6c810b5aab992e54857bc81f5052c2c593c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6f018fd4ec0c6a6f4719b0d547af0cd53cbdffd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6f018fd4ec0c6a6f4719b0d547af0cd53cbdffd4"}]},"branch":"refs/heads/master"},"4a4f4c3c9b42db45da7fe4cb7267ff0d1c8ce29f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-17 10:39:50.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/82/833982/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/82/833982/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/833982/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/833982/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/833982/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/82/833982/2"}}},"commit":{"parents":[{"commit":"7c22039858a5a3045adae149d55138bbfbe6dd2f","subject":"Add coverage for helper methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/7c22039858a5a3045adae149d55138bbfbe6dd2f"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-03-16 10:10:04.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-03-17 10:39:16.000000000","tz":330},"subject":"Add exception coverage for get, get_size, delete","message":"Add exception coverage for get, get_size, delete\n\nThis patch adds coverage for the various exceptions handled and\nraised in get, get_size and delete methods.\nIt also corrects the behavior of _test_cinder_get_size method where\nclient.volumes returned dictionary instead of a MagicMock and it\nworked due to the existing method \"get\" in dictionary which has\nsame name as \"get\" method in cinderclient. The dictionary object\nis replaced with the appropriate MagicMock object in this patch.\n\nChange-Id: If63a6c810b5aab992e54857bc81f5052c2c593c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/4a4f4c3c9b42db45da7fe4cb7267ff0d1c8ce29f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/4a4f4c3c9b42db45da7fe4cb7267ff0d1c8ce29f"}]},"branch":"refs/heads/master"},"4bd0304efbdfb98e2742741af42b483708f48a41":{"kind":"REWORK","_number":3,"created":"2022-03-22 09:08:05.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/82/833982/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/82/833982/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/833982/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/833982/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/833982/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/82/833982/3"}}},"commit":{"parents":[{"commit":"e5a3c9eefa83f173f12ee38804207e94bbdfa544","subject":"Add coverage for helper methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/e5a3c9eefa83f173f12ee38804207e94bbdfa544"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-03-16 10:10:04.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-03-22 09:02:10.000000000","tz":330},"subject":"Add exception coverage for get, get_size, delete","message":"Add exception coverage for get, get_size, delete\n\nThis patch adds coverage for the various exceptions handled and\nraised in get, get_size and delete methods.\nIt also corrects the behavior of _test_cinder_get_size method where\nclient.volumes returned dictionary instead of a MagicMock and it\nworked due to the existing method \"get\" in dictionary which has\nsame name as \"get\" method in cinderclient. The dictionary object\nis replaced with the appropriate MagicMock object in this patch.\n\nChange-Id: If63a6c810b5aab992e54857bc81f5052c2c593c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/4bd0304efbdfb98e2742741af42b483708f48a41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/4bd0304efbdfb98e2742741af42b483708f48a41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
