)]}'
{"id":"openstack%2Fglance_store~844909","triplet_id":"openstack%2Fglance_store~master~I42404fd34b71b977e39eb7888baf60d18bc25931","project":"openstack/glance_store","branch":"master","attention_set":{},"removed_from_attention_set":{"34665":{"account":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"last_update":"2022-06-13 06:39:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I42404fd34b71b977e39eb7888baf60d18bc25931","subject":"Remove unicode literal strings","status":"ABANDONED","created":"2022-06-07 06:05:42.000000000","updated":"2022-06-13 06:39:59.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9c9100bae53dd1f45b55c6606334baa881c44be0","_number":844909,"virtual_id_number":844909,"owner":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-07 07:40:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-07 07:40:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-07 18:41:51.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"}],"messages":[{"id":"82c9d530f7351ce1d08f04770eccb7e38bf32ba1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"date":"2022-06-07 06:05:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e386b0e9a9d2510e94f752c4519d3fdf9613494d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 07:40:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb1329ee0aa74ff398320e814568b0b2 : SUCCESS in 1h 12m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddce52b7fdb6430f87b192af6a8dbb57 : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01c363d31dda499ab6bbb047cc199ddf : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7eadea09ca8745148a2e1fa897f2ee93 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef21856af46747658f3e6be54c297569 : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24977f7cecb9433bbe1a1db16358ad04 : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fac3a3fc48a4a17a5670fad98d1835b : SUCCESS in 3m 35s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/33c9519bca0845b190e5eeb1c1f6441a : SUCCESS in 8m 41s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/76a88d4e91b644c5870a8d1a386b0dde : FAILURE in 1h 22m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"95ac73439215aac03590ebb5ac033281d99853d4","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-06-07 18:41:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c9100bae53dd1f45b55c6606334baa881c44be0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"date":"2022-06-13 06:39:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"43cfca9bd50ce4d90eb0b3a80959489575106eb8","revisions":{"43cfca9bd50ce4d90eb0b3a80959489575106eb8":{"kind":"REWORK","_number":1,"created":"2022-06-07 06:05:42.000000000","uploader":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"ref":"refs/changes/09/844909/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/09/844909/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/844909/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/844909/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/09/844909/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/09/844909/1"}}},"commit":{"parents":[{"commit":"aeee48b561ba0d100663311c3ba94f0d128105aa","subject":"Update python testing as per zed cycle teting runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/aeee48b561ba0d100663311c3ba94f0d128105aa"}]}],"author":{"name":"anguoming","email":"agm_daydayup@foxmail.com","date":"2022-06-07 06:04:04.000000000","tz":480},"committer":{"name":"anguoming","email":"agm_daydayup@foxmail.com","date":"2022-06-07 06:04:04.000000000","tz":480},"subject":"Remove unicode literal strings","message":"Remove unicode literal strings\n\nChange-Id: I42404fd34b71b977e39eb7888baf60d18bc25931\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/43cfca9bd50ce4d90eb0b3a80959489575106eb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/43cfca9bd50ce4d90eb0b3a80959489575106eb8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
