)]}'
{"id":"openstack%2Fglance_store~853018","triplet_id":"openstack%2Fglance_store~master~I71352e75f6415494709d8fb62c8b13f3ea7c6016","project":"openstack/glance_store","branch":"master","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-12-05 09:24:36.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-12-13 06:52:54.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2022-12-13 06:52:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I71352e75f6415494709d8fb62c8b13f3ea7c6016","subject":"Add region_name option to s3 store","status":"MERGED","created":"2022-08-12 18:49:47.000000000","updated":"2022-12-13 06:53:50.000000000","submitted":"2022-12-13 06:52:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853018","meta_rev_id":"2dbd0861aee65f0308fc92c0a108cdfaac638a32","_number":853018,"virtual_id_number":853018,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-13 06:52:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2022-12-05 14:49:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},{"value":2,"date":"2022-12-05 09:24:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},{"value":1,"date":"2022-12-13 05:22:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":18791,"name":"Glance Bot","email":"glancebot@mailinator.com","username":"glancebot","inactive":true},{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}],"reviewers":{"REVIEWER":[{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":18791,"name":"Glance Bot","email":"glancebot@mailinator.com","username":"glancebot","inactive":true},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-12 18:50:07.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-08-12 18:50:07.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2022-08-12 18:50:07.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-08-12 18:50:07.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-08-12 18:50:07.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-08-12 18:50:07.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2022-08-12 18:50:07.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"state":"REVIEWER"},{"updated":"2022-08-12 18:50:07.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":18791,"name":"Glance Bot","email":"glancebot@mailinator.com","username":"glancebot","inactive":true},"state":"REVIEWER"},{"updated":"2022-08-12 20:46:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-16 09:41:03.000000000","updated_by":{"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},"reviewer":{"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},"state":"REVIEWER"}],"messages":[{"id":"6b87392380d2755300036f95d660f0e1b5725768","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-08-12 18:49:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc97e154effd6632213a35f8e19cb500048af7d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 20:46:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/38d909c79fa74fcf9ae3983d05d64228\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1996ac2c70a4e9b8fe26d47e079e66e : SUCCESS in 1h 17m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7784276f1cb4f58acdfe14673a41026 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c91f510b8ee4731a6aa6a19405a974e : FAILURE in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd3c38da05ec44d1acdadb5945280784 : FAILURE in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa5ecb550fc14e119306701ea1eb8424 : FAILURE in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9f24512942a415eb15d7980e1e71f3d : SUCCESS in 3m 41s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/2bdd0c46ec6f47d8ba8ef1b3c45425ec : SUCCESS in 12m 14s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/3f93a7369eff4bf7870e582f08b1a8f5 : FAILURE in 1h 53m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1dbbf886bedc4a44e675e8c35a47e26e2e982833","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-08-12 21:01:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"70e49483a7ff1fc26915afd41c537d8af285fc00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 22:21:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f97848656d5c4c77a90af55052362d6d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/536133d257774709b15814435e56f31f : SUCCESS in 1h 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fde862a5726043daa23d67482cc4df75 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e2cc9a443ef44d9b00b9e449bd1f5f2 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/742d6efa73d540adb16fa53ce153c49d : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01172a859c5b42de9e7defe3b72e3317 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78be2699c1d444269f591f3677898b8b : SUCCESS in 3m 39s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/5396d7edef3a4dd6a3b16f547b649fe4 : SUCCESS in 8m 38s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/8610c5574b9a4a7ba111561ba8e09e83 : SUCCESS in 1h 18m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9a4e3172fa037584e44853a0cc8cd3e00ec46202","author":{"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},"date":"2022-08-16 09:41:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"389a921bb4633fad321d243194bbe4056a8d1280","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-08-16 09:45:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"df0788244e47d6027342af46b7d201ad5c23fb1a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-18 16:30:30.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c19000ee8f312868e332b8564fa3a7287e126b25","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-08-19 09:34:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"404a58a91b320849ac31858ebd6cd461dbde07a6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-19 09:38:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cbf03a15b9ab58328fe5ed178b92b8fcc0900360","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-08-19 09:49:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3c1e09f7e892a9b9f6382e933388704f841d96c7","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-08-23 07:21:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e595e9d54e414d343d67556ba78f1e95124cc419","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-10-05 18:41:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"91c8ecfda1e4be50087e07ab313db784badfbf33","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-10-18 14:38:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a497a2b34c05939226458bf376e71d9e69c962c9","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-12-01 21:30:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3939ec5878b8d937b4ebd93baca33bdeb8572e36","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-12-02 06:02:29.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1ed8df5552f53b52f4d0e2d7979c2fad5f527b9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-12-02 10:46:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d702e753c3c2f7f26f3afdb7f24cc8e91a8960f5","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-12-02 10:48:12.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"989eb87bbacb88cdc661173bdfccc0f2b5fc7f8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 12:18:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e2a5ca19ebc4f399d617904bfd75b75\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e39bbec0fd354982b0583553e575563c : SUCCESS in 1h 29m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c44a846cdb84c2db69212a68de8436a : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dacf60530da64665b87b4f04bf594947 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83901fc9c1bc424fbefe7e77896f4ceb : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab6d6f332da141c88f6af622127672b5 : SUCCESS in 2m 57s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/a21b6ff970454070a75ab1123716ea7c : SUCCESS in 12m 16s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/c8f86d8cd5a54c43be488be8ce11f269 : SUCCESS in 1h 29m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0453c768fddb47200d7a2fec30efc2e3b048de99","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-12-05 09:21:57.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"062f9a085ba77e7f6dc51036b6a75ee902b9a2a8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-12-05 09:24:36.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"965407bbe49b8945307d8e87d27af52987f05e3d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-12-05 14:49:23.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"89095eda1d732f5c84fd4c9ca1b46321e5e957cb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-12-13 05:22:51.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ecf39725e81bf43330c3239e19c0651930780df1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 05:23:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"33f8b49ce3cb9345f64eeecfa36af6c6042faf30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 06:52:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2da28270d3804ef8b5435930aa9d841a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02d19f1cfe2c4dfb8b7ef9fd68cdccee : SUCCESS in 1h 29m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebbab7fd9eaa447991572a1db88eddab : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f044a75624743138a8d971f9a7a4f85 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c606e99b2fb3490c929d6a467edaec06 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12c7420f509a41739e27e93aced6e6bf : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":3},{"id":"af9c98e6630b531f6963cf9cb612bbd931630864","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 06:52:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"2dbd0861aee65f0308fc92c0a108cdfaac638a32","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 06:53:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa47d294f400427288cd0eb22fe228b1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c35e9d96d9d45c9b9cbff800a5a0df4 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2ad4a73e7f4278fe961e8c75f7f9bdbf51382b64","revisions":{"b170430d04cf96fc07c681e590258588d35afaad":{"kind":"REWORK","_number":1,"created":"2022-08-12 18:49:47.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/18/853018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/18/853018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/18/853018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/18/853018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/18/853018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/18/853018/1"}}},"commit":{"parents":[{"commit":"6851cab51a3d5dbf018f755efee8e0e640ac12ab","subject":"Add debug logs to cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6851cab51a3d5dbf018f755efee8e0e640ac12ab"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2022-08-12 18:45:27.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2022-08-12 18:48:15.000000000","tz":120},"subject":"Add region_name option to s3 store","message":"Add region_name option to s3 store\n\nAdd an option to let the operator choose a proper region_name.\n\nSome operator are not using S3 from amazon but from other provider.\nIn such case, the get_s3_location cannot give good region_name.\n\nAdding an option from config seems a better option that relying on\nhardcoded stuff in the code.\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I71352e75f6415494709d8fb62c8b13f3ea7c6016\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/b170430d04cf96fc07c681e590258588d35afaad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/b170430d04cf96fc07c681e590258588d35afaad"}]},"branch":"refs/heads/master"},"3ad5e92f695b1a3a75f538230ac7097adf753ed9":{"kind":"REWORK","_number":2,"created":"2022-08-12 21:01:09.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/18/853018/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/18/853018/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/18/853018/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/18/853018/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/18/853018/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/18/853018/2"}}},"commit":{"parents":[{"commit":"6851cab51a3d5dbf018f755efee8e0e640ac12ab","subject":"Add debug logs to cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6851cab51a3d5dbf018f755efee8e0e640ac12ab"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2022-08-12 18:45:27.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2022-08-12 21:00:58.000000000","tz":120},"subject":"Add region_name option to s3 store","message":"Add region_name option to s3 store\n\nAdd an option to let the operator choose a proper region_name.\n\nSome operator are not using S3 from amazon but from other provider.\nIn such case, the get_s3_location cannot give good region_name.\n\nAdding an option from config seems a better option that relying on\nhardcoded stuff in the code.\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I71352e75f6415494709d8fb62c8b13f3ea7c6016\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3ad5e92f695b1a3a75f538230ac7097adf753ed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3ad5e92f695b1a3a75f538230ac7097adf753ed9"}]},"branch":"refs/heads/master"},"2ad4a73e7f4278fe961e8c75f7f9bdbf51382b64":{"kind":"REWORK","_number":3,"created":"2022-12-02 10:46:55.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/18/853018/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/18/853018/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/18/853018/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/18/853018/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/18/853018/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/18/853018/3"}}},"commit":{"parents":[{"commit":"052c5cc4b4ed494600612018da4482213f4edc5f","subject":"Merge \"Remove logic for Python \u003c\u003d 2.6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/052c5cc4b4ed494600612018da4482213f4edc5f"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2022-08-12 18:45:27.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2022-12-02 10:46:46.000000000","tz":60},"subject":"Add region_name option to s3 store","message":"Add region_name option to s3 store\n\nAdd an option to let the operator choose a proper region_name.\n\nSome operator are not using S3 from amazon but from other provider.\nIn such case, the get_s3_location cannot give good region_name.\n\nAdding an option from config seems a better option that relying on\nhardcoded stuff in the code.\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I71352e75f6415494709d8fb62c8b13f3ea7c6016\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2ad4a73e7f4278fe961e8c75f7f9bdbf51382b64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2ad4a73e7f4278fe961e8c75f7f9bdbf51382b64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
