)]}'
{"id":"openstack%2Fglance_store~860004","triplet_id":"openstack%2Fglance_store~master~Iaba20c01fb4d994cd738a570197f8f6f04f22606","project":"openstack/glance_store","branch":"master","topic":"overflow-error","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-10-21 16:47:43.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-10-12 14:14:59.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"Iaba20c01fb4d994cd738a570197f8f6f04f22606","subject":"[test-only] OverflowError running on 32-bit systems","status":"MERGED","created":"2022-09-30 16:25:14.000000000","updated":"2022-10-21 16:48:56.000000000","submitted":"2022-10-21 16:47:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"860004-overflow-error","meta_rev_id":"e773ad1fdebd6f973d978d92697ea1bd91f57636","_number":860004,"virtual_id_number":860004,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-21 16:47:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"recommended":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2022-10-21 15:32:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2022-10-21 15:32:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2022-10-12 14:14:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2022-10-21 15:32:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-30 18:12:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-05 18:48:39.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-10-21 15:32:41.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2022-10-21 15:32:43.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"77dbc6ba3b04d30560e7ea2e55d779044c4edf4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-09-30 16:25:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30339a8ceeef7ced04cfd93ceb31aa4744315f1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 18:12:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/823fd8d6aee54343b2355fbb3886fc46\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba61a0ebac5f448aa494b77da6a46c1c : SUCCESS in 1h 33m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62dab30b9cc3490888696a81f1d971ed : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86fb6c81e6f44c01b474d847ff59857e : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56ce6125fc844111a4f7d13dddf84b6a : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4c0383f5c7540ec86a1f87223b0e88f : SUCCESS in 3m 03s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/5cd24fcc29ae4a5b98bb8e1297776e34 : SUCCESS in 8m 31s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/57b6893c2964425694dd32c45550b77e : FAILURE in 1h 42m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"79fd38fdb086cd6971a58e3e1e38ff94777039ea","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-10-05 18:48:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b3abf90b5ae5103804603a4a06ea74bbb23277f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-10-12 13:41:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3bf6507b9d2abfdaeedd0ada29c3e318aebabb71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-10-12 13:49:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9d5074d0ac4819c96b6fde7776390e70b5e5001a","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-10-12 14:14:59.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7ed38f02f765193d2e338801a274c15c2f87eb00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 15:15:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6bd4f3e318394e5a979e6fbf8683a463\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f6bb83295b3d49b382324168971450dd : FAILURE in 1h 03m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5980875902dd4471945efee2cd44fd23 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/789e8662e64d429c8276fe7acb01ccfa : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4bb69055cf6a41d7b0fbbc953ffde35c : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/977a94235f174a559c09d90ffc51b909 : SUCCESS in 3m 55s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/fa5878b80b5a47f39e211bcb9fca77d2 : SUCCESS in 7m 54s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/1668e5d49e37493492202861802cc375 : FAILURE in 1h 24m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"95efcb6d19c97cd24ec9f084c689c63f78bbf222","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-10-12 15:44:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f14e0a3ef647a3f520c411aeeec6c0151a91f34b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 17:47:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ca68432e4e741e79b3af1a33b8837fb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb2b833a86a74b85844ea755bd918bbd : SUCCESS in 1h 49m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8801ff0e2b964be5b2ccd5af84591cbf : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77e48f9e74ad4230a4aa67c9f0f289e0 : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/124b5bf0d73c41879ef7e5008e154b52 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc06358a2dc1443e87ffe07efbbbb6c5 : SUCCESS in 3m 12s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/8567981357804a9da8c796ff07d00882 : SUCCESS in 8m 11s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/76720857d2fb4195a923db44b3bf6442 : SUCCESS in 1h 56m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c93a1cadb9686bae88babc17e7c555bd83242f2b","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-10-21 15:32:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"844dbe60a97a9d824c229df533fef91d30ce10b0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-10-21 15:32:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a0034c3b8b73742a313ad98e921e591c9883a7b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 15:33:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d7eb8bece67da668a956b74f4fee1b9735eb8ab9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 16:47:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecb43e18b27f43b8b26bae848ee259f9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a9f2f40672a41cba3e6bc555111bbbe : SUCCESS in 1h 13m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eee1809dfdab463e8cf8e789e974c3b3 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/910ccbe8e5cf4b7d9bcb60f32e6bac56 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b19e6157d61c449da58f2203f3923707 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf9db2ca63b54dec85b278942253b160 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"fefda32e3072f16e2c9896ddd6bae1793c5337fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 16:47:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e773ad1fdebd6f973d978d92697ea1bd91f57636","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 16:48:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e8b476401744d9f961db676ebe74d4c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0229c41426946cabeb3c361d3dda867 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0df64d5af92b46457fd3f13aceb69b4f94b8b53c","revisions":{"c08c10b574a8263f62b93f96bbf714f150ac7f92":{"kind":"REWORK","_number":1,"created":"2022-09-30 16:25:14.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/04/860004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/04/860004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/04/860004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/04/860004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/04/860004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/04/860004/1"}}},"commit":{"parents":[{"commit":"d2dfe692f366bf0ae55532c9509a6fd11fff410a","subject":"Merge \"Switch to 2023.1 Python3 unit tests and generic template name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d2dfe692f366bf0ae55532c9509a6fd11fff410a"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-09-30 16:19:11.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-09-30 16:19:11.000000000","tz":-240},"subject":"[test-only] OverflowError running on 32-bit systems","message":"[test-only] OverflowError running on 32-bit systems\n\nChange I59b1579dc9877668b82d4195431c1 added tests that mock an\ninput buffer, but the mocked return_value for __len__ exceeds the\nallowable maximum value for an index in the CPython implementation\non 32-bit systems.  Fix this by using half the value and allowing\nthe buffer to be called twice during the tests.\n\nChange-Id: Iaba20c01fb4d994cd738a570197f8f6f04f22606\nCloses-bug: #1991406\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/c08c10b574a8263f62b93f96bbf714f150ac7f92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/c08c10b574a8263f62b93f96bbf714f150ac7f92"}]},"branch":"refs/heads/master"},"0df64d5af92b46457fd3f13aceb69b4f94b8b53c":{"kind":"REWORK","_number":2,"created":"2022-10-12 13:49:16.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/04/860004/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/04/860004/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/04/860004/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/04/860004/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/04/860004/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/04/860004/2"}}},"commit":{"parents":[{"commit":"d2dfe692f366bf0ae55532c9509a6fd11fff410a","subject":"Merge \"Switch to 2023.1 Python3 unit tests and generic template name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d2dfe692f366bf0ae55532c9509a6fd11fff410a"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-09-30 16:19:11.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-10-12 13:45:44.000000000","tz":-240},"subject":"[test-only] OverflowError running on 32-bit systems","message":"[test-only] OverflowError running on 32-bit systems\n\nChange I59b1579dc9877668b82d4195431c1 added tests that mock an\ninput buffer, but the mocked return_value for __len__ exceeds the\nallowable maximum value for an index in the CPython implementation\non 32-bit systems.  Fix this by using half the value and allowing\nthe buffer to be called twice during the tests.\n\nChange-Id: Iaba20c01fb4d994cd738a570197f8f6f04f22606\nCloses-bug: #1991406\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0df64d5af92b46457fd3f13aceb69b4f94b8b53c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0df64d5af92b46457fd3f13aceb69b4f94b8b53c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
