)]}'
{"id":"openstack%2Fglance_store~874535","triplet_id":"openstack%2Fglance_store~master~Idb8abfdab9639b5b70ddddc3a76b1f1d24dd035d","project":"openstack/glance_store","branch":"master","topic":"store-dep-validation","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-02-22 22:00:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idb8abfdab9639b5b70ddddc3a76b1f1d24dd035d","subject":"cinder: Disable cinder stores if cinderclient is not installed","status":"MERGED","created":"2023-02-21 08:08:04.000000000","updated":"2023-02-22 22:05:01.000000000","submitted":"2023-02-22 22:00:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874535-store-dep-validation","meta_rev_id":"487d32d98b041c52e46a35cdeaa53cf3a673f52b","_number":874535,"virtual_id_number":874535,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-22 22:00:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-02-21 08:35:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-02-22 16:05:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-02-22 16:05:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-21 08:35:21.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-02-21 10:26:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-22 16:05:38.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"457997784749e9545ccb5dd68d9227b001503c48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-21 08:08:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f711541d6c9bef22d66098225cb189b551c88a71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-21 08:09:25.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"68266266d2f81cb1c53a6a8a02de4eec13555b7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-21 08:32:06.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f89dc87b92a6444fa3cf676622cbae3cffdbea91","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-21 08:32:16.000000000","message":"Topic set to store-dep-validation","accounts_in_message":[],"_revision_number":3},{"id":"92ef3451b93702f8c41c4552c2eb07ec43b154a7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-21 08:35:21.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0a80e7d1679ac0200e6cdff6d90181030387d004","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 10:26:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b40fb03ee4c456cb350ee6b2f4d8ad3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c185f1f99384ee1839e804fae5d7072 : SUCCESS in 1h 39m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13dac30044ca4bab96cb48b9db4345aa : SUCCESS in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65f43c72066f447898cc7e3f80713cb2 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1dc0ae6b78914c749dad570dbd968cbe : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76c7c8b7ded94a6f86e6f1b81719b6ce : SUCCESS in 2m 59s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/92fba56062e246679a5015cc56189b71 : SUCCESS in 8m 51s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/42c433d8fc9a48e989b7a339dc669302 : FAILURE in 1h 49m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7c6a19cccc6a72b9034fffbbbfec46f9e0bd62b1","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-22 16:05:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ab9fff910699055d8be4417066c20f97c82097a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 16:06:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"132e81d0e3c7b7b68c27222381c502b030f818fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 18:16:40.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b079cec7f99e4b758541e2a5fe4296bb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54f943e94db94fbcb6d8437fcca92661 : TIMED_OUT in 2h 04m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e54884978c144644a128d567578d3379 : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a391e2081c7484187e4a1a9fb78fa8c : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05acf0a95aa44cb7a7f5474a250ad4eb : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/475676107cb5478c906f03454845a49e : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":3},{"id":"fa85bb160caa7ae0eb19932a85676470266bf981","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-22 19:06:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"828cd6422ff2c41e24cc47d24efe53d55b9cd7fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 20:48:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdfd3ef66f534eeb998e3ad39fe930fa\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/098b43fc3f354d138120d4207e90756d : SUCCESS in 1h 15m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7c73bace6da465aa34e33e70088c76d : SUCCESS in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d119d79649314d278ac96228942b2393 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c252c5350231467ea5fd856d9b43aeec : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec7a8b5c87374af1a4c5159bde8e80c9 : SUCCESS in 2m 55s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/ca08cdda581f48a694a8e54b9402ab2b : SUCCESS in 8m 14s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/6706a71989724a67a28868dc751f2df1 : SUCCESS in 1h 33m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e44fb5b113f0f402327263ada24979654e12c87e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 20:48:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"633d3fd320ace3103cce9770830e086aceae59ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 22:00:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2e60fc311da4bab8d87c73dc26e4bd4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7515ceba7c1149618e3cb6f604d933ec : SUCCESS in 1h 08m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20104e490a7941bc85b8430954b34cec : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98cff87d99bd4cd2b22367c46d6d0459 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e417e44f10f47b7802f9b96af5440d1 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9c81bafe7684520a98adab05568d569 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":3},{"id":"7d7ab3e0615b75f63d4b4b979830e56a02078d88","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 22:00:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"487d32d98b041c52e46a35cdeaa53cf3a673f52b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 22:05:01.000000000","message":"Patch Set 3:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccc9d5fcdb8a4c9f8ad961151e9e3b94\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e07c1988f4ae4955831e9ef8448eb3c5 : POST_FAILURE in 4m 16s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"572e73ef135106e8b1dd4378db183c565db2cfc9","revisions":{"17649a790303f6906f12dd8cf2933d94e2842c97":{"kind":"REWORK","_number":1,"created":"2023-02-21 08:08:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/874535/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/35/874535/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/35/874535/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/35/874535/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/35/874535/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/35/874535/1"}}},"commit":{"parents":[{"commit":"64e25979a904e468e1e89d8727e9554662596acf","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/64e25979a904e468e1e89d8727e9554662596acf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-21 08:05:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-21 08:05:54.000000000","tz":540},"subject":"Disable cinder backend if cinderclient is not installed","message":"Disable cinder backend if cinderclient is not installed\n\nChange Iec6ac7020f66f9f9728d6cf4452127b8d55a8b2d made cinderclient\noptional but it did not add proper validation while initializing\ncinder backend. This can cause ugly errors in case users tries to\nenable cinder stores without installing cinderclient.\n\nThis ensures glance_store checks whether cinderclient is available and\ndisables all cinder stores in case cinderclient is not available.\n\nChange-Id: Idb8abfdab9639b5b70ddddc3a76b1f1d24dd035d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/17649a790303f6906f12dd8cf2933d94e2842c97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/17649a790303f6906f12dd8cf2933d94e2842c97"}]},"branch":"refs/heads/master"},"05eeab3b74528851ffb34e48837275a3a52d9fef":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-21 08:09:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/874535/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/35/874535/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/35/874535/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/35/874535/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/35/874535/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/35/874535/2"}}},"commit":{"parents":[{"commit":"64e25979a904e468e1e89d8727e9554662596acf","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/64e25979a904e468e1e89d8727e9554662596acf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-21 08:05:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-21 08:09:05.000000000","tz":540},"subject":"Disable cinder store if cinderclient is not installed","message":"Disable cinder store if cinderclient is not installed\n\nChange Iec6ac7020f66f9f9728d6cf4452127b8d55a8b2d made cinderclient\noptional but it did not add proper validation while initializing\na cinder store. This can cause ugly errors in case users tries to\nenable cinder stores without installing cinderclient.\n\nThis ensures glance_store checks whether cinderclient is available and\ndisables all cinder stores in case cinderclient is not available.\n\nChange-Id: Idb8abfdab9639b5b70ddddc3a76b1f1d24dd035d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/05eeab3b74528851ffb34e48837275a3a52d9fef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/05eeab3b74528851ffb34e48837275a3a52d9fef"}]},"branch":"refs/heads/master"},"572e73ef135106e8b1dd4378db183c565db2cfc9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-02-21 08:32:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/874535/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/35/874535/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/35/874535/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/35/874535/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/35/874535/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/35/874535/3"}}},"commit":{"parents":[{"commit":"64e25979a904e468e1e89d8727e9554662596acf","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/64e25979a904e468e1e89d8727e9554662596acf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-21 08:05:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-21 08:32:06.000000000","tz":0},"subject":"cinder: Disable cinder stores if cinderclient is not installed","message":"cinder: Disable cinder stores if cinderclient is not installed\n\nChange Iec6ac7020f66f9f9728d6cf4452127b8d55a8b2d made cinderclient\noptional but it did not add proper validation while initializing\na cinder store. This can cause ugly errors in case users tries to\nenable cinder stores without installing cinderclient.\n\nThis ensures glance_store checks whether cinderclient is available and\ndisables all cinder stores in case cinderclient is not available.\n\nChange-Id: Idb8abfdab9639b5b70ddddc3a76b1f1d24dd035d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/572e73ef135106e8b1dd4378db183c565db2cfc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/572e73ef135106e8b1dd4378db183c565db2cfc9"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
