)]}'
{"id":"openstack%2Fglance_store~874537","triplet_id":"openstack%2Fglance_store~master~I673473aa8b540c084ce3a4934e8be7069f13e77f","project":"openstack/glance_store","branch":"master","topic":"store-dep-validation","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2023-02-21 08:39:00.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-02-22 17:30:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I673473aa8b540c084ce3a4934e8be7069f13e77f","subject":"rbd: Disable rbd stores if libraries are not available","status":"MERGED","created":"2023-02-21 08:31:58.000000000","updated":"2023-02-22 17:32:43.000000000","submitted":"2023-02-22 17:30:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874537-store-dep-validation","meta_rev_id":"dfc0e07952a7897abe05cc7f13becb024bbf38f8","_number":874537,"virtual_id_number":874537,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-22 17:30:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-02-21 08:39:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-02-22 16:05:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-02-22 16:05:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-21 08:33:58.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2023-02-21 08:39:00.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-02-21 10:38:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-22 16:05:13.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"65bcc0642391a8a6341ec267e458efae2e067c41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-21 08:31:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff86b1d7ea6da8a45e40872cca80d6e40f3e399d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-21 08:32:27.000000000","message":"Topic set to store-dep-validation","accounts_in_message":[],"_revision_number":1},{"id":"e581bda28e306cfb38313f2e0c45793d8463879d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-21 08:33:58.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f945636e4b21c9e8143789ca80781bf7ac2d4900","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-21 08:37:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ba8783f851e35f6cf78bf6e10c53e92b82ec10c1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-21 08:38:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0e9972e175ac7f440cfb532d477cc1d03c83e40","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-21 08:39:00.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"44f36ee221a7d89dce0c99ea1bd94d620f28a199","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 10:38:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b44a0012c7c4768af0253bca5a9af34\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7727ec4735b34f21bd1c3d0286dd33c4 : SUCCESS in 1h 28m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22fcd43b217b461dab4eeff7acefbfe6 : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bbc23d0697443d29013e4bbe4c9f828 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03c8774732e44beb89551b5ddb6b4dea : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7ca8178f5dc402e86029cae78368959 : SUCCESS in 3m 34s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/478a0ef351b34d15aa671d502fb81c4d : SUCCESS in 8m 29s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/ca2943802a7b4ffbad69bfca3e45b567 : FAILURE in 1h 53m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7ce7d046be6de68b3c89ba895f17a91fcd0fdf16","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-22 16:05:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"67ca9ba28b4825a49c51dacf0b9782a282ff5a13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 16:05:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"18d43e7cb7e3e4034cbe138938756be5f705f26a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 17:30:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dd629c5acf741a5ab7453fb1d9d843a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41c3df5c1d1e4cddbcbd8b4d56a9dee1 : SUCCESS in 1h 19m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30d5cfd16d794dafb296896d06727efc : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ba412960b724def8e8032fead68376c : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2f8c13aac3b4bce84a1323ddd9b5ca4 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20c7eb0f3a384d89949b65d0ef521f2d : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":2},{"id":"bc32fe10d0dc424340af4bc625267a6648da74c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 17:30:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"dfc0e07952a7897abe05cc7f13becb024bbf38f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 17:32:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcc96662d2da4b1e8a5ae03301c5c230\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9339109b41ef46df9236eac0150f33ca : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ea5bab9b577b9ff2c9cf1a14f21a0a4873379212","revisions":{"0b26abf1b9852e2d7fa076b1ce14d1a6e8d79d4b":{"kind":"REWORK","_number":1,"created":"2023-02-21 08:31:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/874537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/37/874537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/37/874537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/37/874537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/37/874537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/37/874537/1"}}},"commit":{"parents":[{"commit":"64e25979a904e468e1e89d8727e9554662596acf","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/64e25979a904e468e1e89d8727e9554662596acf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-21 08:30:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-21 08:31:43.000000000","tz":540},"subject":"rbd: Disable rbd stores if libraries are not available","message":"rbd: Disable rbd stores if libraries are not available\n\nThis ensures glance_store checks whether the dependent libraries (rbd\nand rados) when adding rbd store and disables all rados stores in case\nthese libraries are not available.\n\nThis avoid ugly error (eg. AttributeError) caused by missing libraries.\n\nChange-Id: I673473aa8b540c084ce3a4934e8be7069f13e77f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0b26abf1b9852e2d7fa076b1ce14d1a6e8d79d4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0b26abf1b9852e2d7fa076b1ce14d1a6e8d79d4b"}]},"branch":"refs/heads/master"},"ea5bab9b577b9ff2c9cf1a14f21a0a4873379212":{"kind":"REWORK","_number":2,"created":"2023-02-21 08:37:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/874537/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/37/874537/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/37/874537/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/37/874537/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/37/874537/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/37/874537/2"}}},"commit":{"parents":[{"commit":"64e25979a904e468e1e89d8727e9554662596acf","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/64e25979a904e468e1e89d8727e9554662596acf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-21 08:30:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-21 08:37:48.000000000","tz":540},"subject":"rbd: Disable rbd stores if libraries are not available","message":"rbd: Disable rbd stores if libraries are not available\n\nThis ensures glance_store checks whether the dependent libraries (rbd\nand rados) when adding rbd store and disables all rados stores in case\nthese libraries are not available.\n\nThis avoid ugly error (eg. AttributeError) caused by missing libraries.\n\nChange-Id: I673473aa8b540c084ce3a4934e8be7069f13e77f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ea5bab9b577b9ff2c9cf1a14f21a0a4873379212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ea5bab9b577b9ff2c9cf1a14f21a0a4873379212"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
