)]}'
{"id":"openstack%2Fglance_store~890532","triplet_id":"openstack%2Fglance_store~master~Ifbef97eb95e6400ec7efa1058791e6f0f90e4459","project":"openstack/glance_store","branch":"master","hashtags":[],"change_id":"Ifbef97eb95e6400ec7efa1058791e6f0f90e4459","subject":"test CI","status":"ABANDONED","created":"2023-08-04 15:07:40.000000000","updated":"2023-08-31 07:20:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"1b0343dddb94f683fbc830e900731cffd5cf88a1","_number":890532,"virtual_id_number":890532,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-08-04 17:07:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-08-04 17:07:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3dbefa1a0bbf59539962fddb11448c87f3fbe10c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2023-08-04 15:07:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c46bae19457e50edafa6de613f480a7221accccb","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2023-08-04 15:07:55.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"72fa42f6d110b68a92c5fcb10b3503aae9f46eef","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2023-08-04 15:08:07.000000000","message":"Topic dummy removed","accounts_in_message":[],"_revision_number":1},{"id":"48e850a8eb8663a3572caf46a3675fbbf0a65353","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 17:07:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d694413c548742a9ae7e4fe325c3a9c6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/93dc3c66c36940a981c05c5655355563 : SUCCESS in 2m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/90aa3b288d9e4ddb92c0d88cb7058329 : SUCCESS in 2m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a50b68882e124387894530df4d9e5ce5 : SUCCESS in 1h 11m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e1a99c76ed54e9bbddc9da2a656f8e4 : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1123ab08efd542eeac36e4a29bf0cdbd : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d2742225b694823954276e5fa2d9be2 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/329822a3f5ee430e816574ba40b803e9 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df66c7029db04e36aef5cb9fa09735f5 : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a160d49e0d464c228bc0c97e568aa36b : SUCCESS in 5m 41s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/9413192a835e4dce94f64145b7379b35 : SUCCESS in 9m 41s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/e19b4cd6a8a74de38205062066b09e4d : SUCCESS in 1h 57m 21s","accounts_in_message":[],"_revision_number":1},{"id":"1b0343dddb94f683fbc830e900731cffd5cf88a1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2023-08-31 07:20:08.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed2fbbb226e6bee2cea0139e5e1f4cfd8b84a805","revisions":{"ed2fbbb226e6bee2cea0139e5e1f4cfd8b84a805":{"kind":"REWORK","_number":1,"created":"2023-08-04 15:07:40.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/32/890532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/32/890532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/32/890532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/32/890532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/32/890532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/32/890532/1"}}},"commit":{"parents":[{"commit":"0c60291637d1c941dcd8d2e022acb22ba0bed440","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0c60291637d1c941dcd8d2e022acb22ba0bed440"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2023-08-04 15:07:26.000000000","tz":180},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2023-08-04 15:07:26.000000000","tz":180},"subject":"test CI","message":"test CI\n\nChange-Id: Ifbef97eb95e6400ec7efa1058791e6f0f90e4459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ed2fbbb226e6bee2cea0139e5e1f4cfd8b84a805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ed2fbbb226e6bee2cea0139e5e1f4cfd8b84a805"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
