)]}'
{"id":"openstack%2Fglance_store~893696","triplet_id":"openstack%2Fglance_store~master~I57fc8cbd6308c4a82a638a2d1e997605736aedc4","project":"openstack/glance_store","branch":"master","topic":"deprecate-vmware","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-12-07 15:12:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I57fc8cbd6308c4a82a638a2d1e997605736aedc4","subject":"Deprecate VMWare Datastore","status":"MERGED","created":"2023-09-05 03:26:07.000000000","updated":"2023-12-07 15:13:26.000000000","submitted":"2023-12-07 15:12:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893696-deprecate-vmware","meta_rev_id":"2a0b73974be439f6eeecbd09688f15a754be2cd6","_number":893696,"virtual_id_number":893696,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-07 15:12:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2023-12-07 13:05:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2023-12-07 13:05:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-05 06:02:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-07 13:05:02.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"3960634d4a9eddd862e9d9cc9aa72addf6ce4df2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-05 03:26:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec9d50cc7fe5285f93ed9125880dd13383e46f82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 06:02:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b6e0583be6b4e0786d085e2bf0d7f3a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/44e757ddb69e41f8931a2aad9813feaf : SUCCESS in 2m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f0af14db2e74041b680ce225002b46b : SUCCESS in 1h 08m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5187dcf3a22d47679d68d51c042fb32e : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fad20506a2b54e1380904366cbb4866e : FAILURE in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d872d1d832374c6789e64592423cd9ed : FAILURE in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9af2c816f114a3fb539dbfa2e00313d : FAILURE in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb6e59efc3e04a01a9db7f1dbefe50f8 : FAILURE in 3m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e24ec490c9c4b5f83654e28a073b450 : SUCCESS in 2m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b22b89520ada4fc0ae3e95f0989a7ca1 : SUCCESS in 3m 42s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/c6fb2970395f47cf8676994e3a4d249b : SUCCESS in 8m 19s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/fba14d8c1439430eae064667159732fb : SUCCESS in 2h 30m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2d72caa01f3b1e46f3f37fe4714337cc07058bb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-05 13:02:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"13ade634419d4795afeb1cf26849d377bd992291","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 15:34:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40b5abaaa54b46bc80f4b4c24e49d953\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6bfbf45cc40744209a5ccefa5000309e : SUCCESS in 1h 07m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c694489c5db42679460288fec22777f : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/227660b03943461088eb71636da5d714 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28cd689c16af4da986aeaec3686ea0b0 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c63e60ebbdbb4300afb81e04ef4fb47a : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36f44eeee3bd4f6b834736cbf566d01c : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a679657859f540739a075b7bc4f850e6 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/496745974ded42e6b5d6b64847353401 : SUCCESS in 4m 12s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/fb807008ffbd473bb55c318cbaa72406 : SUCCESS in 8m 08s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/05f86efef0934bdc8118a8de57daaf55 : SUCCESS in 2h 30m 53s","accounts_in_message":[],"_revision_number":2},{"id":"c9b25734cab59788dce00a769262da6965e5178a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-11 13:40:59.000000000","message":"Topic set to deprecate-vmware","accounts_in_message":[],"_revision_number":2},{"id":"1e025715aa4051c27e6932b48df506b364f77847","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-12-07 13:05:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c243a7a6b4cd61700dd4cf822c69f60621d4c093","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 13:05:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"753939638c123c6fd0d3f75c73a491a6da40d738","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 15:12:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c78005e1396476ea14d14cd4ee9287b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/645d76ba85dd45688c2135c35d1170cf : SUCCESS in 1h 58m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2914c27c846b48b9bb66d1a09bb2a90c : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71a48aeaf9a845238868f6c5d6f8587e : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/639f8a9c783142a391e49e94f37b5bd1 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fe2517090b2441c92dc258ee647e6c7 : SUCCESS in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e025f4e53a59416eb9c83076e8eff9c6 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"ce7b7ccc7d7ddc89ab6fb46369c3ff3f7339de2c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 15:12:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2a0b73974be439f6eeecbd09688f15a754be2cd6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 15:13:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1573d2a46d64d0d88e396f753c408eb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa4c47279d12443bbc2b9081b91f3961 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d4f7e52ee734ddc9d49f657f2956c98 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ea61721442ee692ca9a3ac8e76024b0fa2324006","revisions":{"fc27dd85bec42e186afe08ceea7a8619712c56dc":{"kind":"REWORK","_number":1,"created":"2023-09-05 03:26:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/893696/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/96/893696/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/893696/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/893696/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/893696/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/96/893696/1"}}},"commit":{"parents":[{"commit":"054bd5ddf5d4d255076bd5f44296f2521e899394","subject":"Merge \"RBD: Wrap RBD calls in native threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/054bd5ddf5d4d255076bd5f44296f2521e899394"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-05 03:23:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-05 03:25:58.000000000","tz":540},"subject":"Deprecate VMWare Datastore","message":"Deprecate VMWare Datastore\n\nThe vmwareapi virt driver in nova was marked as experimental some\ncycles ago due to lack of CI and maintainers[1]. This deprecates\nthe VMWare Datastore so that we can remove it when nova removes\nthe virt driver. The store driver is used along with the vmwareapi virt\ndriver and we don\u0027t expect any usage of this store driver with\nthe other virt drivers such as libvirt driver.\n\n[1] 713ef7839670bd5d73af52c211a14932adc933ba\n\nChange-Id: I57fc8cbd6308c4a82a638a2d1e997605736aedc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/fc27dd85bec42e186afe08ceea7a8619712c56dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/fc27dd85bec42e186afe08ceea7a8619712c56dc"}]},"branch":"refs/heads/master"},"ea61721442ee692ca9a3ac8e76024b0fa2324006":{"kind":"REWORK","_number":2,"created":"2023-09-05 13:02:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/893696/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/96/893696/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/893696/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/893696/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/893696/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/96/893696/2"}}},"commit":{"parents":[{"commit":"054bd5ddf5d4d255076bd5f44296f2521e899394","subject":"Merge \"RBD: Wrap RBD calls in native threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/054bd5ddf5d4d255076bd5f44296f2521e899394"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-05 03:23:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-05 13:02:19.000000000","tz":540},"subject":"Deprecate VMWare Datastore","message":"Deprecate VMWare Datastore\n\nThe vmwareapi virt driver in nova was marked as experimental some\ncycles ago due to lack of CI and maintainers[1]. This deprecates\nthe VMWare Datastore so that we can remove it when nova removes\nthe virt driver. The store driver is used along with the vmwareapi virt\ndriver and we don\u0027t expect any usage of this store driver with\nthe other virt drivers such as libvirt driver.\n\n[1] 713ef7839670bd5d73af52c211a14932adc933ba\n\nChange-Id: I57fc8cbd6308c4a82a638a2d1e997605736aedc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ea61721442ee692ca9a3ac8e76024b0fa2324006"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ea61721442ee692ca9a3ac8e76024b0fa2324006"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
