)]}'
{"id":"openstack%2Fglance_store~893980","triplet_id":"openstack%2Fglance_store~master~I852beab63512f2469f92181a2ba427cbaa172ec2","project":"openstack/glance_store","branch":"master","topic":"bug/2030825","attention_set":{},"removed_from_attention_set":{"34437":{"account":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"last_update":"2024-11-04 18:03:45.000000000","reason":"\u003cGERRIT_ACCOUNT_34437\u003e replied on the change","reason_account":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-11-11 20:53:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I852beab63512f2469f92181a2ba427cbaa172ec2","subject":"S3: add option to specify a custom CA cert bundle","status":"MERGED","created":"2023-09-06 20:50:48.000000000","updated":"2024-11-11 20:54:29.000000000","submitted":"2024-11-11 20:53:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893980-bug/2030825","meta_rev_id":"d775da45c093e4cf3dec427d513a840bc2efd430","_number":893980,"virtual_id_number":893980,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-11 20:53:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"all":[{"value":1,"date":"2024-11-04 18:03:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},{"value":2,"date":"2024-11-04 18:02:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2024-11-11 15:15:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},{"value":1,"date":"2024-11-11 15:39:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-06 22:40:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-31 12:17:26.000000000","updated_by":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"reviewer":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"state":"REVIEWER"},{"updated":"2024-10-31 14:37:36.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"CC"},{"updated":"2024-11-04 18:02:13.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-11-11 15:15:55.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"}],"messages":[{"id":"f17a0b7b3aff4217192eeed6817eef39f719e4ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-09-06 20:50:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18d0aae56e0fd5e569e92ad3f24e3d9254f52acb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 22:40:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5bbc9065225f4606aad0bb10e5ec63ba\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0f8390d35fa4bd8861700e3144e4841 : SUCCESS in 1h 42m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f88c1f4ee72c4b148180c867d6eaff84 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b746e799de34353887cdb26afa0dbc0 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/415590b48b804f94a62c1aba8d3b6629 : SUCCESS in 6m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e0e97a7d194b4c8585205533d9f8b847 : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32e8d858e38b46149641f020ffe9e03f : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d8345758f3a4002b9b7b4317a1fd71e : SUCCESS in 5m 19s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/f614441ec28c4b7dbf55e3e6f43c789c : SUCCESS in 10m 23s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/bc0d554ae07c45abaede7bcea6859342 : FAILURE in 1h 44m 57s","accounts_in_message":[],"_revision_number":1},{"id":"d344ffaeb37b2344a7a6032797af664a07aebcd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-10-09 11:31:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f082fd2fc1311e5027381da82c4704a8c5aaa079","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-10-09 11:31:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"31ea557f7965afc1960038cfff21dd5b02f6f851","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 14:17:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31405ce38f7f4080982354f41b43c650\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05ca1710ce244a709c03a6aedbfae5e7 : SUCCESS in 1h 43m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/debb89ab956d4e6c8bd81f650cf18890 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c84fe22aae9743f1ba45e77805986b15 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a88e1e5a33ff459ea540d85fdea533f3 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4941bfd41da4198afe360ae992a21f4 : SUCCESS in 2m 43s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/efe9955c7e76438a903f8f5141d064d6 : SUCCESS in 7m 21s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/0996b559f39c4416ae6127871597096f : SUCCESS in 2h 39m 52s","accounts_in_message":[],"_revision_number":2},{"id":"8ea54ba4e0f1dd3368ee5595f633ff6913fe36ba","author":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"date":"2024-10-31 12:17:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4907ddfcd15c85f3a6e7604f962bc18197b025cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-11-04 16:57:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8b90a2bb3f12050f8795ffe1169753e6b74bdeff","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-04 18:02:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c331247284c6b50741e715c60da4042b92d34692","author":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"date":"2024-11-04 18:03:45.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7f9da81e9424b90588cff0d9dc7df02665b3ffa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 19:20:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66e646b832274f1981a8537c26d08d6f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28436565c716421984ebae1d74f6ae08 : SUCCESS in 1h 56m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/850c917ea5ed4be0b01c621712116d21 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/969a22b6f23245359528432746a88e97 : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6152c1ef563344e2a4a3d5e8d3216271 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad9a61e55e434045a4db19682c8c6783 : SUCCESS in 3m 15s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/96433ae94ddb4b56b8b234313b27265e : SUCCESS in 8m 46s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/4478a7e2c3ab4e6da0a7f38ef2648e21 : FAILURE in 2h 22m 05s","accounts_in_message":[],"_revision_number":3},{"id":"aaf5683b14662ab73cbc64798b3b4a4804245490","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-08 05:52:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4aa7b5364aeb219a0fd4bcc639a397db7c1bccba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 08:29:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f78e8b0218d54e32bd87f49b451c2aa7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c446227b1ae849cb8ddd5df46eb59e3c : SUCCESS in 1h 29m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6ffe2db14794b71bcdba7bde35d76d7 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/350eecab00aa455c884e85d0161cef59 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95d6f2a0672b478398b7d2e298ab3f6c : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18fc124d23204c5487f60bb9d2fb338c : SUCCESS in 3m 06s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/9e0e71f2eb78427d81d57dc02b7ede6f : SUCCESS in 8m 19s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/6721114a441b4e1d8719a7b6a9bc455f : SUCCESS in 2h 35m 35s","accounts_in_message":[],"_revision_number":3},{"id":"8c1b6b07015e21d972dadbe817ddfa0e52da59db","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2024-11-11 15:15:55.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"757cae5813a72d52aed0911b41e351c6b145db07","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-11 15:39:45.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"28323058894f7d151adb94e82df593cb20c74201","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 15:40:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"68345108009490f6144a79eb6742d92ba0e1a073","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 17:06:48.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b08f0fa8582846228ff48a08a6b1c008\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b0c5ab2cf344414b907a6ac3f1674c4 : FAILURE in 1h 26m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f9379c9d9d94817b0275168c4232c6b : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3df97d7025c145999c7d819f4cf4fa0d : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f34f1afc80b4494bb1ad0d16c85f423e : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e968b1cc46c4895b9d739052d9a8b78 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":3},{"id":"6099df2da8a0e744e8d9ee5f9588e3fa0b0800c2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-11 17:11:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"358eb0d90713295d99ffc45d2be71e5da519b012","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 19:12:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46fff2ca559a4cc89c679aebf1c7f123\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c43473d361434f3098e01c9ea883568e : SUCCESS in 1h 23m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7ebbf724774432a86225434a66346a3 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39327920f45b4581bfe6e397928cd64d : SUCCESS in 5m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ea36ec1b4844584bd14bbd1cc3dd83f : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d22702f13a124cffb0d1d6e4540281d1 : SUCCESS in 3m 00s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/00ccd3a1db2f4e329f59673520f0ca32 : SUCCESS in 9m 30s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/4e48a8c913974f1f89bdeaae98b169dd : SUCCESS in 1h 59m 27s","accounts_in_message":[],"_revision_number":3},{"id":"d826a700ad5d6f67830b7e71c06c0caca9916541","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 19:12:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b72eeb092b807a8f4892dcf116ce8cfb5800ef38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 20:53:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eca9846a41b4b2e9a3b5cecad15c470\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd1e0439171f4d7a989f4b5da2e5a32f : SUCCESS in 1h 35m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85892d413be9407cbdeb8c74ed67b0b4 : SUCCESS in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3ef050390a147a699fd2ab221ebf47e : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1951ed72e4644e5e91f4dfe0d4e66feb : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d0a1d331cd6412b84b82fa58351776e : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":3},{"id":"2381fdbd6e2017a474bc16d108ad524c01dd0361","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 20:53:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d775da45c093e4cf3dec427d513a840bc2efd430","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 20:54:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c09f6983f3da42a5ab00e08c15ea87ee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d59852c309a42aebd468d63ab5d070a : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f9db33edc78411a0d7cf82a19b01da24ad7304fd","revisions":{"d1419d973971cf6b3ee4771786207378bd1a0f4c":{"kind":"REWORK","_number":1,"created":"2023-09-06 20:50:48.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/80/893980/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/80/893980/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/80/893980/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/80/893980/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/80/893980/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/80/893980/1"}}},"commit":{"parents":[{"commit":"0c60291637d1c941dcd8d2e022acb22ba0bed440","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0c60291637d1c941dcd8d2e022acb22ba0bed440"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-09-06 20:08:55.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-09-06 20:47:05.000000000","tz":120},"subject":"[WIP] S3: add option to specify a custom CA cert bundle","message":"[WIP] S3: add option to specify a custom CA cert bundle\n\nUsers can:\n- set this option to the path to their CA cert bundle;\n- use the default value to use the default CA cert bundle used by\n  botocore.\n\nWe do not allow users to disable SSL certificates verification.\n\nCloses-Bug: #2030825\n\nChange-Id: I852beab63512f2469f92181a2ba427cbaa172ec2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d1419d973971cf6b3ee4771786207378bd1a0f4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d1419d973971cf6b3ee4771786207378bd1a0f4c"}]},"branch":"refs/heads/master"},"0651e1671577afd8428c4576b0da2278e0e28c5e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-09 11:31:32.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/80/893980/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/80/893980/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/80/893980/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/80/893980/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/80/893980/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/80/893980/2"}}},"commit":{"parents":[{"commit":"23916fca593760c445cfcbfa8e335fc61f730dd8","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/23916fca593760c445cfcbfa8e335fc61f730dd8"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-09-06 20:08:55.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-10-09 11:31:32.000000000","tz":0},"subject":"[WIP] S3: add option to specify a custom CA cert bundle","message":"[WIP] S3: add option to specify a custom CA cert bundle\n\nUsers can:\n- set this option to the path to their CA cert bundle;\n- use the default value to use the default CA cert bundle used by\n  botocore.\n\nWe do not allow users to disable SSL certificates verification.\n\nCloses-Bug: #2030825\n\nChange-Id: I852beab63512f2469f92181a2ba427cbaa172ec2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0651e1671577afd8428c4576b0da2278e0e28c5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0651e1671577afd8428c4576b0da2278e0e28c5e"}]},"branch":"refs/heads/master","description":"Rebase"},"f9db33edc78411a0d7cf82a19b01da24ad7304fd":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2024-11-04 16:57:38.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/80/893980/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/80/893980/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/80/893980/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/80/893980/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/80/893980/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/80/893980/3"}}},"commit":{"parents":[{"commit":"086bc2058fe95165b93a92a983a03466e4bdbbb2","subject":"Move to python3.9 as minimal python version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/086bc2058fe95165b93a92a983a03466e4bdbbb2"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-09-06 20:08:55.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-11-04 16:57:21.000000000","tz":60},"subject":"S3: add option to specify a custom CA cert bundle","message":"S3: add option to specify a custom CA cert bundle\n\nUsers can:\n- set this option to the path to their CA cert bundle;\n- use the default value to use the default CA cert bundle used by\n  botocore.\n\nWe do not allow users to disable SSL certificates verification.\n\nCloses-Bug: #2030825\n\nChange-Id: I852beab63512f2469f92181a2ba427cbaa172ec2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/f9db33edc78411a0d7cf82a19b01da24ad7304fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/f9db33edc78411a0d7cf82a19b01da24ad7304fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
