)]}'
{"id":"openstack%2Fglance_store~894514","triplet_id":"openstack%2Fglance_store~master~Ib370f527c06dc85bcfd9df6ca1efb2a4e8cb5e7d","project":"openstack/glance_store","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-01-19 16:44:23.000000000","reason":"Change was submitted"},"18002":{"account":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"last_update":"2024-01-19 16:44:23.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-01-19 04:43:41.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-19 16:44:23.000000000","reason":"Change was submitted"},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2024-01-19 16:44:23.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-01-19 15:01:21.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2024-01-19 16:44:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib370f527c06dc85bcfd9df6ca1efb2a4e8cb5e7d","subject":"RBD: Use rados_connect_timeout to override timeout","status":"MERGED","created":"2023-09-11 10:06:39.000000000","updated":"2024-01-19 16:45:51.000000000","submitted":"2024-01-19 16:44:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"894514","meta_rev_id":"8491a12825ac9fa74b02010d025d066f5bff9f4c","_number":894514,"virtual_id_number":894514,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-19 16:44:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-01-19 04:43:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2024-01-19 15:01:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2024-01-19 15:01:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-11 11:38:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-11 15:20:32.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2023-09-11 15:21:44.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-09-11 18:11:10.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2023-09-12 01:22:44.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-09-12 02:20:57.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2023-09-19 04:47:24.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"},{"updated":"2023-12-07 13:36:22.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"329eb134d752c58e436699fdc6c45dc015ca84e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-11 10:06:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52b82ed51248183dd7be24801e0c22b0d5139565","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 11:38:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d874c23a75d4beeb2d992f38641c439\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e503c906790243db9ec34901afd50921 : SUCCESS in 1h 26m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d1c4640fe3a410cb699d0c6daa8e232 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a5aae1f6e8c4fc29a05228f0191a1ba : FAILURE in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2362acb62724c1b9c7883b51f0e5de5 : FAILURE in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf6d7b5ed5c742e19961fe26fde9ad99 : FAILURE in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab28afc52ad74940b31af3400ee6eb51 : FAILURE in 3m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3158ec3f02c844e99b09db0aac725a5d : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/888e11ac56bc40bf80096c95ed8fa99b : SUCCESS in 3m 26s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/d9eb797799674d799bd6fb33e4c58b99 : SUCCESS in 9m 53s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/9ec4d3ea22c34bec9ced6182cfa267b5 : FAILURE in 23m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7d7ca0e3f983c33705089075008d786c46265a3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-11 12:30:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bb7e3b106971b93703afe204b5d1c4cdbf687025","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 14:27:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f578b3eedf740919c8acbb5dfe54d04\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93fe47e4a4f44a87a1cd30e313a23fbe : SUCCESS in 1h 14m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff9e19f98d2f4699b7a6f6cfb35fd302 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ced58cb4cf58430cbdd636a32022e0d3 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9579b52089254fa682da6a16a1a2fbfa : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/130f08cb8f5c4214a8e74d567d236e55 : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18a874a96033426b9ce2421f6aa4cd35 : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7decfa46718f479d8b25f68bc6c79f7e : SUCCESS in 3m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85a64552752a4440acb1328a221ef962 : SUCCESS in 3m 01s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/664c4969f8674956aa893d85fc35102d : SUCCESS in 8m 55s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/080f56870212405998ac073f13d68b1e : SUCCESS in 1h 55m 20s","accounts_in_message":[],"_revision_number":2},{"id":"7fb492ab0e520464117f1e60245b165e6cc35c88","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-09-11 15:20:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"52131ba1ed6362f74f1f57aa9616705a4d4c5718","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-09-11 15:21:44.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"60371539e2817ead1a63a676543ea9ad187be5a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-11 15:45:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"eb05755af830f6b3a4e8f4a7287ba5b028b1169b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-11 15:48:19.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"76d683dbcccfd5fce87f5c30af87f20360400c51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 18:06:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f88bc0eb10f6420196c89b50156f66cb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09fb2b670f8b41c8bd1a6ce09baa9292 : SUCCESS in 1h 42m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0334d8f7dc8346929e9d44b3a01df60c : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/378fe4abf4b14521bf739be513d55168 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65ae9725c34042cda24e00c2c6508c8b : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d251087b9134781870602dff5f5bd7a : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6be65636958640a9a5a6e08487f76d51 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fe700a2ba3a4ea4a56887831cf6f795 : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e238b81ce474f74b51d212f410fe747 : SUCCESS in 4m 04s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/4a21b8414db9436e9eb785fe5597fbe4 : SUCCESS in 9m 26s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/34803a3d2c8b4fe3b110f5ef48677846 : SUCCESS in 2h 19m 27s","accounts_in_message":[],"_revision_number":3},{"id":"25d8c4864802bfbeadcee0b4eb18dbaf1dda800e","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2023-09-11 18:11:10.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"471e8059da05addf72fc2d6f28ae0cb03811831f","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-09-12 02:20:57.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2f84cf0b85289a3a60fdcf4de74eda123e5d0f58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-27 02:33:20.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"df42f8f5ddde5c3b0f0a8ccc6b969180eede6661","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-27 03:28:20.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"142d61d71afce40595866593633d57a1b4b66015","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 06:04:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/87596496f1684237a467a1cf3b333c8f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6fad6d708311490bbf9d929bda65946c : SUCCESS in 1h 19m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cddc5fedfac4849ae1d54599f58f14f : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/213ef8e84bab443aa1f97e2c855b95bd : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f798161129554324bbf8112edcef05d1 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b629ad9058f1487f831aa4e12acf77db : SUCCESS in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13ea7b8094414a2e9364a70b09beccef : SUCCESS in 4m 25s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/eaef6567a47244088874abbf00b84e78 : SUCCESS in 9m 03s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/1014b72ad4bb4b4d8c2aec35a2b9820b : TIMED_OUT in 2h 34m 50s","accounts_in_message":[],"_revision_number":5},{"id":"aa5d400a70310ffc51e3fa173dda31d5140d59a2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-27 06:15:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"105667eb29446c96ab3246ec36a2038a35f81061","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 08:55:22.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0063c65a6ccc4ae19f091fcbece874f6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/494e20ba05d540db99de49cbd8666d51 : SUCCESS in 1h 26m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd98e55270c74ffd9f2aed988f189776 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/666e8b038ed74db2aea729bf8d9d1a7d : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2adaa233eca4328ba8d8cc13ce19858 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90e7d91382d547f0ad394563f7be022a : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7397b4ea1ff438697c38fa31967c787 : SUCCESS in 3m 48s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/abdb3fb13d664d29a91aacb3a6a560bc : SUCCESS in 9m 30s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/a36f4f39812f4349a810d00d3c839882 : TIMED_OUT in 2h 33m 31s","accounts_in_message":[],"_revision_number":5},{"id":"a91bcad57d31e27776e3d9ad8bc4e7eef0ef28ed","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-11-29 15:55:33.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f241344dbd7bd735f1e6aff31c4fe8e4d79ae05a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-03 06:06:03.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"67183ce6cbe4406f266f00af54ee65677857595c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-12-07 13:36:22.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d3db0f8ad9a9867676566feec6c1eb67dc7b2d70","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-08 03:46:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b5018f23d6f971f19c5d14ddd0cb34a7be92a502","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-13 08:25:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7537f3620d8af9f2a7f8611d167ccf908d55ee6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 10:58:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6214f8a99c484e84b0167681ee05c564\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a91d1da856bd496e92f646e103214894 : SUCCESS in 1h 57m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c70bb982fec4d6fa3447dd0c24bd16b : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7712de0a65b343a0a3a0a914c37efd52 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/149e78f14ed74ab7a995a98cb0f1a6d0 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5a5f78130d144f799611a0fb552fa4e : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09c93d46f3a44f6d887d82b78e1fac8d : SUCCESS in 3m 10s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/4fd1f49de51342a19298a249b1bb00f6 : SUCCESS in 7m 10s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/62c5520f6d9d44a8ac5a647cc08c16a4 : SUCCESS in 2h 31m 40s","accounts_in_message":[],"_revision_number":5},{"id":"bb2219dab267da8a92e2cca41ad1b177a01f0d3c","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2023-12-13 20:20:30.000000000","message":"Patch Set 5: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"7aaf70b5049c56fbef98be314bae366cd337d20b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-01-18 17:02:52.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"e3d344c795a6c95ebfc20d6320b1a4ae7411eedd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-18 17:26:03.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"50e7896f73b22707b5003b72d2ec19edb6b75f6a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-18 17:29:12.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"93095ae6dfb69d0f336ab4cd6718f4be35d49cd0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-01-18 18:12:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cf78b6c6f6d2d8397fd6c32d8b4daa668367ea03","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-18 18:25:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"88de99cd03678b22eb354491d7a3abc580f6a559","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-19 04:17:08.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"870c9e716a1878cb421b0e24fa7dd9d98439f219","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-19 04:18:54.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"86cb983008acd3483dcfcf60ce65e2cad77ba9ec","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-19 04:43:41.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"be80afbf4528a2f26c34dc5538aa5a8534625615","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 06:57:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ba5db2d1ea047b493b3f17593b2b0d3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52fc235e3ca84c989220babb6c6ddc98 : SUCCESS in 1h 47m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1441868b85df44f687a8bbc8c061d924 : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33b38f20c409413e8b7dde9cfe7fe7b8 : SUCCESS in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9909a9a2d66f4e91bb9bb28632f863ee : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fde53c9944b498995c4f6e9aa2df6f8 : SUCCESS in 2m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce87be77daad487896e9f41177a9ac8b : SUCCESS in 2m 57s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/10718c69289e4f449d52dd23690dbd9d : SUCCESS in 7m 58s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/85ff95714a3e434ea75915e5a3ef1022 : TIMED_OUT in 2h 33m 56s","accounts_in_message":[],"_revision_number":6},{"id":"d197ee549db8fb1cebf540707895d9f968a0729a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-19 07:27:46.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1bf922978ac6bea87aa0a50224e89a0da2450927","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-19 08:10:22.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cd77cd68f8a7d34a97f479bd11f454d97943a1e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 10:14:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb707b009d474f2c8c357c1abe606aee\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce92e8cf431f4833849d03482f414534 : SUCCESS in 1h 13m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64dbb9c7a94b4292a20b666480cedd9e : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15de343dc5a24476a768fe0d81402c2b : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/181da6b40b0d4b6b8974889efedbe4c6 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c758984bf5d74a9898684d3283d044dd : SUCCESS in 2m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fdc483c05444f6d863057f4b959e4dc : SUCCESS in 3m 05s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/7a81d2dad09e452c99a31e2ec0817a65 : SUCCESS in 8m 25s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/fc3c7e6c36094de7a298aab9e0539031 : SUCCESS in 2h 44m 04s","accounts_in_message":[],"_revision_number":6},{"id":"1de0d04fdac806a4738927cfc70efdee01449170","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-01-19 15:01:21.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"450d268c95f91946c15166ddedfc0469b82ca27a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 15:01:53.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"fe27c03b53db6fc9215d4f669abbd89a9945cf91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 16:44:23.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9401a760aca34c3291e1ca829ccb1d10\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6db07f43e33244418fda945310ec3aa7 : SUCCESS in 1h 37m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a7e48c1f7a84c1fbc167d8a5c5e1920 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbb8c1daa1c14c0193b6263cdd9df935 : SUCCESS in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b095bc4694b4b8ca12b08ee5620c9a7 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93b141781c7d428a802e19bb00f1bde1 : SUCCESS in 3m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14311cef113143d1b594964670612ea2 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":6},{"id":"b7f5434b0caaed39ce677a84baaf28928b10bc65","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 16:44:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"8491a12825ac9fa74b02010d025d066f5bff9f4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 16:45:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90511e4843894afbbb8c2ecca9daecb8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec744a7add8f4850859a40d264782069 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/984da30a7dd342c8acb2cce5fccb9dde : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c197bbdccadf799be2d86da28f1d2302ccf5e1e8","revisions":{"a69ef33784278ea3f9083642b29c57171b9162a7":{"kind":"REWORK","_number":1,"created":"2023-09-11 10:06:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/894514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/14/894514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/894514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/894514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/894514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/14/894514/1"}}},"commit":{"parents":[{"commit":"0c60291637d1c941dcd8d2e022acb22ba0bed440","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0c60291637d1c941dcd8d2e022acb22ba0bed440"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-11 10:01:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-11 10:05:47.000000000","tz":540},"subject":"RBD: Use rados_connect_timeout to override timeout","message":"RBD: Use rados_connect_timeout to override timeout\n\nThis imports the existing rados_connect_timeout handling in cinder so\nthat users can use a different timeouts for Glance specifically,\ninstead of relying on the global ceph.conf options.\n\nChange-Id: Ib370f527c06dc85bcfd9df6ca1efb2a4e8cb5e7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a69ef33784278ea3f9083642b29c57171b9162a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a69ef33784278ea3f9083642b29c57171b9162a7"}]},"branch":"refs/heads/master"},"2179cdd7390fabf2cbcf8b30112e29f19d4c5e88":{"kind":"REWORK","_number":2,"created":"2023-09-11 12:30:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/894514/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/14/894514/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/894514/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/894514/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/894514/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/14/894514/2"}}},"commit":{"parents":[{"commit":"0c60291637d1c941dcd8d2e022acb22ba0bed440","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0c60291637d1c941dcd8d2e022acb22ba0bed440"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-11 10:01:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-11 12:30:50.000000000","tz":540},"subject":"RBD: Use rados_connect_timeout to override timeout","message":"RBD: Use rados_connect_timeout to override timeout\n\nThis imports the existing rados_connect_timeout handling in cinder so\nthat users can use a different timeouts for Glance specifically,\ninstead of relying on the global ceph.conf options.\n\nChange-Id: Ib370f527c06dc85bcfd9df6ca1efb2a4e8cb5e7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2179cdd7390fabf2cbcf8b30112e29f19d4c5e88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2179cdd7390fabf2cbcf8b30112e29f19d4c5e88"}]},"branch":"refs/heads/master"},"dee18a016581c4614f0dc4fdd93e0d470bd7c927":{"kind":"REWORK","_number":3,"created":"2023-09-11 15:45:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/894514/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/14/894514/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/894514/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/894514/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/894514/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/14/894514/3"}}},"commit":{"parents":[{"commit":"0c60291637d1c941dcd8d2e022acb22ba0bed440","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0c60291637d1c941dcd8d2e022acb22ba0bed440"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-11 10:01:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-11 15:45:28.000000000","tz":540},"subject":"RBD: Use rados_connect_timeout to override timeout","message":"RBD: Use rados_connect_timeout to override timeout\n\nThis imports the existing rados_connect_timeout option handling in\ncinder so that users can use a different timeouts for Glance\nspecifically, instead of relying on the global ceph.conf options.\n\nThis parameter was initially deprecated in Zed release by [1] and has\nhad no effect since then, but this change restores the parameter with\nthe logic to override the timeout value in RADOS client.\n\n[1] b1d0feeba4e16046c7840bd4966f31a0e7527e67\n\nRelated-Bug: #1983499\nChange-Id: Ib370f527c06dc85bcfd9df6ca1efb2a4e8cb5e7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/dee18a016581c4614f0dc4fdd93e0d470bd7c927"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/dee18a016581c4614f0dc4fdd93e0d470bd7c927"}]},"branch":"refs/heads/master"},"1c9accaf4b678421cf2606c25b5162cb2860b691":{"kind":"NO_CHANGE","_number":4,"created":"2023-11-27 02:33:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/894514/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/14/894514/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/894514/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/894514/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/894514/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/14/894514/4"}}},"commit":{"parents":[{"commit":"0c60291637d1c941dcd8d2e022acb22ba0bed440","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0c60291637d1c941dcd8d2e022acb22ba0bed440"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-27 02:33:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-27 02:33:12.000000000","tz":540},"subject":"RBD: Use rados_connect_timeout to override timeout","message":"RBD: Use rados_connect_timeout to override timeout\n\nThis imports the existing rados_connect_timeout option handling in\ncinder so that users can use a different timeouts for Glance\nspecifically, instead of relying on the global ceph.conf options.\n\nThis parameter was initially deprecated in Zed release by [1] and has\nhad no effect since then, but this change restores the parameter with\nthe logic to override the timeout value in RADOS client.\n\n[1] b1d0feeba4e16046c7840bd4966f31a0e7527e67\n\nRelated-Bug: #1983499\nChange-Id: Ib370f527c06dc85bcfd9df6ca1efb2a4e8cb5e7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/1c9accaf4b678421cf2606c25b5162cb2860b691"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/1c9accaf4b678421cf2606c25b5162cb2860b691"}]},"branch":"refs/heads/master"},"ff6466dc1df24d50cf16d8e3b456955b2d533299":{"kind":"REWORK","_number":5,"created":"2023-11-27 03:28:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/894514/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/14/894514/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/894514/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/894514/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/894514/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/14/894514/5"}}},"commit":{"parents":[{"commit":"0c60291637d1c941dcd8d2e022acb22ba0bed440","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0c60291637d1c941dcd8d2e022acb22ba0bed440"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-27 02:33:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-27 03:28:12.000000000","tz":540},"subject":"RBD: Use rados_connect_timeout to override timeout","message":"RBD: Use rados_connect_timeout to override timeout\n\nThis imports the existing rados_connect_timeout option handling in\ncinder so that users can use a different timeouts for Glance\nspecifically, instead of relying on the global ceph.conf options.\n\nThis parameter was initially deprecated in Zed release by [1] and has\nhad no effect since then, but this change restores the parameter with\nthe logic to override the timeout value in RADOS client.\n\n[1] b1d0feeba4e16046c7840bd4966f31a0e7527e67\n\nRelated-Bug: #1983499\nChange-Id: Ib370f527c06dc85bcfd9df6ca1efb2a4e8cb5e7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ff6466dc1df24d50cf16d8e3b456955b2d533299"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ff6466dc1df24d50cf16d8e3b456955b2d533299"}]},"branch":"refs/heads/master"},"c197bbdccadf799be2d86da28f1d2302ccf5e1e8":{"kind":"REWORK","_number":6,"created":"2024-01-19 04:17:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/894514/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/14/894514/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/894514/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/894514/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/894514/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/14/894514/6"}}},"commit":{"parents":[{"commit":"0c60291637d1c941dcd8d2e022acb22ba0bed440","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0c60291637d1c941dcd8d2e022acb22ba0bed440"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-27 02:33:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-19 04:16:24.000000000","tz":540},"subject":"RBD: Use rados_connect_timeout to override timeout","message":"RBD: Use rados_connect_timeout to override timeout\n\nThis imports the existing rados_connect_timeout option handling in\ncinder so that users can use a different timeouts for Glance\nspecifically, instead of relying on the global ceph.conf options.\n\nThis parameter was initially deprecated in Zed release by [1] and has\nhad no effect since then, but this change restores the parameter with\nthe logic to override the timeout value in RADOS client.\n\n[1] b1d0feeba4e16046c7840bd4966f31a0e7527e67\n\nRelated-Bug: #1983499\nChange-Id: Ib370f527c06dc85bcfd9df6ca1efb2a4e8cb5e7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/c197bbdccadf799be2d86da28f1d2302ccf5e1e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/c197bbdccadf799be2d86da28f1d2302ccf5e1e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
