)]}'
{"id":"openstack%2Fglance_store~906087","triplet_id":"openstack%2Fglance_store~master~I9193df38d613259b61bb369fa1040fb2c51a21d7","project":"openstack/glance_store","branch":"master","topic":"accesskey","attention_set":{"36707":{"account":{"_account_id":36707,"name":"lujiefsi","display_name":"lujiefsi","email":"lujie@ict.ac.cn","username":"lujiefsi"},"last_update":"2024-01-20 17:39:59.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9193df38d613259b61bb369fa1040fb2c51a21d7","subject":"Do not show access_key in s3 driver","status":"MERGED","created":"2024-01-19 05:19:12.000000000","updated":"2024-01-20 17:39:59.000000000","submitted":"2024-01-19 15:03:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"906087-accesskey","meta_rev_id":"d1b1b15e3891afb8b0b3fffcda3245f3b9b3c007","_number":906087,"virtual_id_number":906087,"owner":{"_account_id":36707,"name":"lujiefsi","display_name":"lujiefsi","email":"lujie@ict.ac.cn","username":"lujiefsi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-19 15:03:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2024-01-19 10:20:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2024-01-19 09:37:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-01-19 14:56:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2024-01-19 13:37:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2024-01-19 14:56:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-19 07:57:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-19 09:37:55.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-01-19 10:20:45.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2024-01-19 14:56:09.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-01-20 17:39:59.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"78cd24fedf65056dfb8ed3cc27477bfc45c01425","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36707,"name":"lujiefsi","display_name":"lujiefsi","email":"lujie@ict.ac.cn","username":"lujiefsi"},"date":"2024-01-19 05:19:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73a3b0c3bb6904cfa04493da106021a75b6deeeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 07:57:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ae2649d2ffc49178dc69688cc596847\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3969f6cfb0f6420c8d56b50cc19dad2f : SUCCESS in 1h 31m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b03a1f7ee1042fdac6977b13ccdeb64 : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c77e08d892c416aaf2263776dec428e : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/90d3e7800e52427f8de7c5f806076649 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cd49dadd9044e0f9dde7a8d81dce045 : SUCCESS in 3m 01s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/4fe87b659d8f4db6ac69c57b5c8856ef : SUCCESS in 8m 29s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/45cf8e7d08024105a704dcfff37a0c29 : TIMED_OUT in 2h 33m 00s","accounts_in_message":[],"_revision_number":1},{"id":"028e1805ddf3d821519226577add6f8c186631ec","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-19 09:37:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"197b8d3f29894d9c2176dc9ea25ee6574ac5080b","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-01-19 10:20:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1cb579e73b57e36161c23352ff256fb05b84e62e","author":{"_account_id":36707,"name":"lujiefsi","display_name":"lujiefsi","email":"lujie@ict.ac.cn","username":"lujiefsi"},"date":"2024-01-19 10:23:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"812d9f80df3c77dd05ef3e16150f96f2636b29f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 12:46:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d57d29ea83f6420ba70135822c1ad0e6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bca9709857ca4cc5be163abc68b1be75 : SUCCESS in 1h 28m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/299192c0165540b0a48f5984531aa2f3 : SUCCESS in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72963728a6514418aed9fa229b481e81 : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a7b1c4bc35242d3b6311c66c2c206e8 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8deea14daeb46d98ce95e87ddf61246 : SUCCESS in 2m 46s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/6eb34b4c0c2b4ee39b2c9a3643ecfcd3 : SUCCESS in 8m 40s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/f24fcc28f09343888fc01b59b9996d2a : SUCCESS in 3h 02m 31s","accounts_in_message":[],"_revision_number":1},{"id":"830a5116a55b439eb409e361b7af2327dd6af75a","author":{"_account_id":36707,"name":"lujiefsi","display_name":"lujiefsi","email":"lujie@ict.ac.cn","username":"lujiefsi"},"date":"2024-01-19 13:30:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c54a6f3b308f1695ccfbabd3c6c27e2d6c7f30d0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-19 13:37:43.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ade17166694504e068323b98a3861e4a5973a682","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 13:38:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85b9d42041a95455bb2cb19a900ac057fc34d19f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-01-19 14:56:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7f089f9ff52645af0f7d545dede75d5429f488d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 15:03:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/440e6b857ba04e4281572bf75d64bb85\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6879b79504e4f38ba6a31922565e5a3 : SUCCESS in 1h 20m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1199ba2a6d5540e2bd8583aaf3ab8d3d : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf60e468b0474ff68cc18c9f4e029489 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6735682d2ace46798db7c091d3370cd4 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91ffbfa088684d7e963e5bb6c42ecad5 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"92a00c0d037250ea9967d67199c0e3e9f8d07321","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 15:03:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5d7b54b89a19355df0480dc1559b4719f0e59fd9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 15:04:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6bd85739d7e4e2dbf3eec16679183db\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1b1d7d17f254340af7effbe114291e6 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1},{"id":"d1b1b15e3891afb8b0b3fffcda3245f3b9b3c007","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-20 17:39:59.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a5ba027922ba1230b4ae9abb810f36427be6354a","revisions":{"a5ba027922ba1230b4ae9abb810f36427be6354a":{"kind":"REWORK","_number":1,"created":"2024-01-19 05:19:12.000000000","uploader":{"_account_id":36707,"name":"lujiefsi","display_name":"lujiefsi","email":"lujie@ict.ac.cn","username":"lujiefsi"},"ref":"refs/changes/87/906087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/87/906087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/87/906087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/87/906087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/87/906087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/87/906087/1"}}},"commit":{"parents":[{"commit":"7b1df2a651dfece4794c56d8914c7290f65fb8b8","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/7b1df2a651dfece4794c56d8914c7290f65fb8b8"}]}],"author":{"name":"lujie","email":"lujie@ict.ac.cn","date":"2024-01-19 05:12:20.000000000","tz":480},"committer":{"name":"lujie","email":"lujie@ict.ac.cn","date":"2024-01-19 05:12:20.000000000","tz":480},"subject":"Do not show access_key in s3 driver","message":"Do not show access_key in s3 driver\n\nCloses-Bug: #2047688\nChange-Id: I9193df38d613259b61bb369fa1040fb2c51a21d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a5ba027922ba1230b4ae9abb810f36427be6354a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a5ba027922ba1230b4ae9abb810f36427be6354a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
