)]}'
{"id":"openstack%2Fglance_store~906885","triplet_id":"openstack%2Fglance_store~stable%2F2023.1~Ib4c900e14788c6b581ce277a71c02d7b07369e6a","project":"openstack/glance_store","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-04-30 11:12:23.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-05-20 18:30:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib4c900e14788c6b581ce277a71c02d7b07369e6a","subject":"Increase timeout of glance_store-src-ceph-tempest","status":"MERGED","created":"2024-01-29 04:49:57.000000000","updated":"2024-05-20 18:32:42.000000000","submitted":"2024-05-20 18:30:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906885","cherry_pick_of_change":906489,"cherry_pick_of_patch_set":1,"meta_rev_id":"7910c34b59f165ea5f92828bdaf960b04424891d","_number":906885,"virtual_id_number":906885,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-20 18:30:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-04-30 11:12:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-04-30 11:11:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-05-20 17:00:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-29 04:49:57.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-01-29 04:50:27.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-29 04:50:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-30 11:11:52.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"0eaa05c7ce4ab54ed45de53d5526bf71fd8c0280","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-29 04:49:57.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.\n\nThe following files contain Git conflicts:\n* .zuul.yaml","accounts_in_message":[],"_revision_number":1},{"id":"ba10a29b672fd9c1962a121636550bef9db565b6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-01-29 04:50:27.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance_store on branch stable/2023.1.  The error was:\n\n  while parsing a block collection\n  in \"openstack/glance_store/.zuul.yaml@stable/2023.1\", line 1, column 1\ndid not find expected \u0027-\u0027 indicator\n  in \"openstack/glance_store/.zuul.yaml@stable/2023.1\", line 133, column 1","accounts_in_message":[],"_revision_number":1},{"id":"ee3f2dd19489e9f5c59bc3955e30abe5b0acf164","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 04:50:31.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/glance_store on branch stable/2023.1.  The\nproblem was:\n\n  while parsing a block collection\n    in \"openstack/glance_store/.zuul.yaml@stable/2023.1\", line 1, column 1\n  did not find expected \u0027-\u0027 indicator\n    in \"openstack/glance_store/.zuul.yaml@stable/2023.1\", line 133, column 1","accounts_in_message":[],"_revision_number":1},{"id":"83385625483d9f1ae5d497d821f46cf0f0ab2ff2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-29 05:49:01.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c1bd9098323afea92503e4e820ebae219510cd97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 07:41:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bd1e302e8b54f52bd311546286c75e1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e01c2843a72f4f1784dbd67c02c2e147 : SUCCESS in 1h 25m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b41031ee63d4e32a1e702d02aba5af6 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e5bd096ebb943d9975c113390785941 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95b5a3a5abf6404b8ade19fbd1dfe099 : SUCCESS in 3m 48s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/da8dc121414a44109bbbfa9e17e7e61a : SUCCESS in 8m 14s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/0d4f8c34efeb457a8a6f27f0ad82f9a5 : SUCCESS in 1h 44m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ee7ae619edb212671780d9d694413e37f1f1fa70","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-04-30 11:11:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f2cdb35363c5613d4b638e699be130a5ced70224","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-04-30 11:12:23.000000000","message":"Patch Set 2: Code-Review+2\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"b03366d9b0d8d092cc451bfb966a5837152b5747","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-20 17:00:04.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aa89a99012a16e425e6f2a5d7c78ed2daa33e16d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 17:00:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8c8677fe4617aa899344ff59eccdccad7a695fcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 18:30:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/341c92f81c2046d5a460cd390acacc1b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1176cd1612ad4583ac2ceb6197e0594e : SUCCESS in 1h 30m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e74688a64fd0442899ae0d3aba69d102 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19bcb33c15d141fdb773c5b0c64d6507 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36d527950ee2413584c923713e527f3d : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":2},{"id":"90baf710d3046bfaabdf1c9eed19eb2fc20b6d7f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 18:30:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7910c34b59f165ea5f92828bdaf960b04424891d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 18:32:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df7cafdfd465407fb69a75e4ca1bfa6a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d0c002535394dd6be58d16a6443e915 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"59bd7a530f91369e13489381473ca5e71972f418","revisions":{"fddd45d0528fbbb620fac66afa69718d3f6e67c6":{"kind":"REWORK","_number":1,"created":"2024-01-29 04:49:57.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/85/906885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/85/906885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/85/906885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/85/906885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/85/906885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/85/906885/1"}}},"commit":{"parents":[{"commit":"aae8199f82f173a57e73e5068c2b682f3d372e6a","subject":"RBD: Trash image when snapshots prevent deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/aae8199f82f173a57e73e5068c2b682f3d372e6a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-04 07:46:25.000000000","tz":540},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-29 04:49:57.000000000","tz":0},"subject":"Increase timeout of glance_store-src-ceph-tempest","message":"Increase timeout of glance_store-src-ceph-tempest\n\nSince [1] was merged, tempest takes quite long time because of lower\nconcurrency. However it\u0027s not feasible to increase the concurrency\nagain because of resource limitation.\n\nThis increases timeout as a short-term solution to unblock gate.\n\n[1] https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/865315\n\nChange-Id: Ib4c900e14788c6b581ce277a71c02d7b07369e6a\n(cherry picked from commit c2e2e62cfa5f29a0f619dc9afa0aad83179533cd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/fddd45d0528fbbb620fac66afa69718d3f6e67c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/fddd45d0528fbbb620fac66afa69718d3f6e67c6"}]},"branch":"refs/heads/stable/2023.1"},"59bd7a530f91369e13489381473ca5e71972f418":{"kind":"REWORK","_number":2,"created":"2024-01-29 05:49:01.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/85/906885/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/85/906885/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/85/906885/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/85/906885/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/85/906885/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/85/906885/2"}}},"commit":{"parents":[{"commit":"aae8199f82f173a57e73e5068c2b682f3d372e6a","subject":"RBD: Trash image when snapshots prevent deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/aae8199f82f173a57e73e5068c2b682f3d372e6a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-04 07:46:25.000000000","tz":540},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-29 05:48:47.000000000","tz":0},"subject":"Increase timeout of glance_store-src-ceph-tempest","message":"Increase timeout of glance_store-src-ceph-tempest\n\nSince [1] was merged, tempest takes quite long time because of lower\nconcurrency. However it\u0027s not feasible to increase the concurrency\nagain because of resource limitation.\n\nThis increases timeout as a short-term solution to unblock gate.\n\n[1] https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/865315\n\nChange-Id: Ib4c900e14788c6b581ce277a71c02d7b07369e6a\n(cherry picked from commit c2e2e62cfa5f29a0f619dc9afa0aad83179533cd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/59bd7a530f91369e13489381473ca5e71972f418"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/59bd7a530f91369e13489381473ca5e71972f418"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
