)]}'
{"id":"openstack%2Fglance_store~907734","triplet_id":"openstack%2Fglance_store~stable%2F2023.1~I8dc564bed33d6fc71965f4f573ae9109b410b1d4","project":"openstack/glance_store","branch":"stable/2023.1","topic":"accesskey/stable-2023.1","hashtags":[],"change_id":"I8dc564bed33d6fc71965f4f573ae9109b410b1d4","subject":"s3: Do not log access keys","status":"ABANDONED","created":"2024-02-05 13:49:51.000000000","updated":"2024-02-15 22:35:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0d9dd2aac3eada88def774396a7e8d1bb62a95c6","_number":907734,"virtual_id_number":907734,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-05 15:49:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-05 13:49:51.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-02-05 15:49:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7bf38158fd48269cc96060c1322f263df24942a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-05 13:49:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"064812d4c40b5a3af5beedc751c29addac16f131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 15:49:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a73ab3fcfd0c4706b183fd0365d94021\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b644c912eac84697a1025f505c048602 : SUCCESS in 1h 05m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e37eecc74894ba7bc409a312e3d3b81 : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/900057b1edbb448e8402ad1b5370ad08 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f646143884074ec4b1118ff221ccade0 : SUCCESS in 3m 26s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/1b0163dfc1544dc3a5a86e277ab163db : SUCCESS in 11m 33s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/ce5bc78433cf41659a8ee8857f3f094a : SUCCESS in 1h 52m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d9dd2aac3eada88def774396a7e8d1bb62a95c6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-15 22:35:14.000000000","message":"Abandoned\n\nNo longer needed; see https://review.opendev.org/c/openstack/glance_store/+/906484","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5dd046ce6ebdb5dbf85bb96dca860314fad76a84","revisions":{"5dd046ce6ebdb5dbf85bb96dca860314fad76a84":{"kind":"REWORK","_number":1,"created":"2024-02-05 13:49:51.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/34/907734/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/34/907734/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/34/907734/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/34/907734/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/34/907734/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/34/907734/1"}}},"commit":{"parents":[{"commit":"d5d38fa0a9fe4a07935e4e903997666b952f91a7","subject":"Do not show access_key in s3 driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d5d38fa0a9fe4a07935e4e903997666b952f91a7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-20 17:09:05.000000000","tz":540},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-02-05 13:48:45.000000000","tz":60},"subject":"s3: Do not log access keys","message":"s3: Do not log access keys\n\nThe previous attempt a5ba027922ba1230b4ae9abb810f36427be6354a was\nincomplete and there are still a few more logs where access keys are\nlogged. This fixes these to avoid leaking access keys to log.\n\nRelated-Bug: #2047688\nChange-Id: I8dc564bed33d6fc71965f4f573ae9109b410b1d4\n(cherry picked from commit d6e531af4821c8466b1e9404f12f89f6216417f2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/5dd046ce6ebdb5dbf85bb96dca860314fad76a84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/5dd046ce6ebdb5dbf85bb96dca860314fad76a84"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
