)]}'
{"id":"openstack%2Fglance_store~907737","triplet_id":"openstack%2Fglance_store~stable%2Fzed~I8dc564bed33d6fc71965f4f573ae9109b410b1d4","project":"openstack/glance_store","branch":"stable/zed","topic":"accesskey/zed","hashtags":[],"change_id":"I8dc564bed33d6fc71965f4f573ae9109b410b1d4","subject":"s3: Do not log access keys","status":"ABANDONED","created":"2024-02-05 13:52:44.000000000","updated":"2024-02-15 22:35:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d84090bae182796eab52fff1e2734971152ece8a","_number":907737,"virtual_id_number":907737,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-05 15:51:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-05 13:52:44.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-02-05 15:51:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"81084ef0bb12517fc6592ebd86810880b21378c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-05 13:52:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08ffbb6338bf00310fa4c33a3060c0bde98a4410","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 15:51:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5869af07d004f329e5fdd05e6d6c22e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4e25eaa561d43bebba7b9f88389fe46 : SUCCESS in 1h 02m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b2c264b759e4d48b931dae5ae712462 : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b3c42fdb45445f18c3cd16010d5b304 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71ab603969c143e29ff1e583594b098a : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b9d03d0e6a64e85aa4f8278dc1c6851 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/493d88027bc4483faf828d6372b9e4a5 : SUCCESS in 3m 05s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/1d1d0998b7674ed3b9508c205815cac4 : SUCCESS in 8m 22s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/c3c662477c57483ba0df67c8284f8492 : SUCCESS in 1h 56m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d84090bae182796eab52fff1e2734971152ece8a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-15 22:35:47.000000000","message":"Abandoned\n\nNo longer needed; see https://review.opendev.org/c/openstack/glance_store/+/906484","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"028ff2aa543144a240c910e7f9dbcb1ad692e83e","revisions":{"028ff2aa543144a240c910e7f9dbcb1ad692e83e":{"kind":"REWORK","_number":1,"created":"2024-02-05 13:52:44.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/37/907737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/37/907737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/37/907737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/37/907737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/37/907737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/37/907737/1"}}},"commit":{"parents":[{"commit":"5a03d0a13349654e6e39562acfa9218e27dd3afd","subject":"Do not show access_key in s3 driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/5a03d0a13349654e6e39562acfa9218e27dd3afd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-20 17:09:05.000000000","tz":540},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-02-05 13:52:31.000000000","tz":60},"subject":"s3: Do not log access keys","message":"s3: Do not log access keys\n\nThe previous attempt a5ba027922ba1230b4ae9abb810f36427be6354a was\nincomplete and there are still a few more logs where access keys are\nlogged. This fixes these to avoid leaking access keys to log.\n\nRelated-Bug: #2047688\nChange-Id: I8dc564bed33d6fc71965f4f573ae9109b410b1d4\n(cherry picked from commit d6e531af4821c8466b1e9404f12f89f6216417f2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/028ff2aa543144a240c910e7f9dbcb1ad692e83e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/028ff2aa543144a240c910e7f9dbcb1ad692e83e"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
