)]}'
{"id":"openstack%2Fglance_store~907739","triplet_id":"openstack%2Fglance_store~stable%2Fyoga~I8dc564bed33d6fc71965f4f573ae9109b410b1d4","project":"openstack/glance_store","branch":"stable/yoga","topic":"accesskey/yoga","hashtags":[],"change_id":"I8dc564bed33d6fc71965f4f573ae9109b410b1d4","subject":"s3: Do not log access keys","status":"ABANDONED","created":"2024-02-05 13:54:47.000000000","updated":"2024-02-08 21:12:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"795f0088681430bc025125edbedd3579c016a8d0","_number":907739,"virtual_id_number":907739,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-05 15:53:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-05 13:54:47.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-02-05 15:53:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e4a63cf106cdfa53083f2acddb27c7cf39ecb20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-05 13:54:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ff1a260a2263bcbf6ff555acf145e8450063b82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 15:53:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a85564fc573e4d7b8ff8dcc4093a0832\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b941534c6d774b8a92c6d67b7f24eeba : SUCCESS in 1h 05m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9475ea69e6e4b05bdbe66bb172bc299 : SUCCESS in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2138f46c97b4e178dc275ed02a44df2 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ce897c6b06a46fe917d71ad90d2166b : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2707d59ffe79423f936c000c10d9c414 : SUCCESS in 2m 54s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/629d1a29f720415d884af263215dd748 : SUCCESS in 8m 18s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/7da4651ae6f44a03abc40b7a69a7de1d : FAILURE in 1h 52m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"795f0088681430bc025125edbedd3579c016a8d0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-08 21:12:19.000000000","message":"Abandoned\n\nAs discussed at today\u0027s Glance meeting [0], this patch will be abandoned and re-proposed to unmaintained/yoga\n\n[0] https://meetings.opendev.org/meetings/glance/2024/glance.2024-02-08-14.00.log.html","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40a851b3bcedb28a7e692622d0fc42ffc43388b0","revisions":{"40a851b3bcedb28a7e692622d0fc42ffc43388b0":{"kind":"REWORK","_number":1,"created":"2024-02-05 13:54:47.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/39/907739/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/39/907739/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/39/907739/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/39/907739/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/39/907739/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/39/907739/1"}}},"commit":{"parents":[{"commit":"fe8c218f6b93d6fb27a7377fea5339562a02ad8a","subject":"Do not show access_key in s3 driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/fe8c218f6b93d6fb27a7377fea5339562a02ad8a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-20 17:09:05.000000000","tz":540},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-02-05 13:54:35.000000000","tz":60},"subject":"s3: Do not log access keys","message":"s3: Do not log access keys\n\nThe previous attempt a5ba027922ba1230b4ae9abb810f36427be6354a was\nincomplete and there are still a few more logs where access keys are\nlogged. This fixes these to avoid leaking access keys to log.\n\nRelated-Bug: #2047688\nChange-Id: I8dc564bed33d6fc71965f4f573ae9109b410b1d4\n(cherry picked from commit d6e531af4821c8466b1e9404f12f89f6216417f2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/40a851b3bcedb28a7e692622d0fc42ffc43388b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/40a851b3bcedb28a7e692622d0fc42ffc43388b0"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
