)]}'
{"id":"openstack%2Fglance_store~911005","triplet_id":"openstack%2Fglance_store~master~I4f27a9800f239da8dbf29f4c028678df1f867664","project":"openstack/glance_store","branch":"master","topic":"new-location-api-support","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-04-04 14:23:02.000000000","reason":"\u003cGERRIT_ACCOUNT_19138\u003e replied on the change","reason_account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-04-09 06:30:07.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-04-22 15:46:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4f27a9800f239da8dbf29f4c028678df1f867664","subject":"Use normal credentials for legacy image update","status":"MERGED","created":"2024-03-04 22:06:34.000000000","updated":"2024-04-22 15:47:22.000000000","submitted":"2024-04-22 15:46:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911005-new-location-api-support","meta_rev_id":"56d4f516024758f1d78aa0ca63faf0d0db3f687d","_number":911005,"virtual_id_number":911005,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-22 15:46:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2024-04-22 10:31:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2024-04-09 06:30:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-04-22 14:00:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2024-04-22 14:00:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-04 23:42:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-05 05:23:55.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-03-05 12:48:15.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2024-04-22 14:00:32.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"ec4b347f67774778d24520870ab9e97d877eb06a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-04 22:06:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db445986752b69e22b7281c0dba60a9eb4874a48","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-04 22:07:11.000000000","message":"Topic set to new-location-api-support","accounts_in_message":[],"_revision_number":1},{"id":"c4ab8ad40a694c13631828ab11ecfa76417b12be","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-04 22:32:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b474affdd1d90d2a338ab65da806649c7b3306f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 23:42:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b81418eed254423aad6600af30447c6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5fd88247555246dca4dfb8354481f29f : SUCCESS in 1h 35m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6717628ca2e544b3bd36df5f23fd2f1a : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7eadf91ff354a24befa51bf4a5f4d89 : FAILURE in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f77b57fc5772422590017a5647ce282f : FAILURE in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd760efa5642403f8aaa4fb64e137257 : SUCCESS in 2m 56s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/2e7df9f71d954db8a42c5a5f9caefc71 : SUCCESS in 7m 45s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/fc6d6ac2168e4a7482a6bbee51d306d0 : FAILURE in 23m 06s","accounts_in_message":[],"_revision_number":1},{"id":"650ce534508425df0bff53cb40db106a3212ff39","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-03-05 05:23:55.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"973c75c5c43ed49f462f2f399b5e22ef0f779f55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-05 12:19:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4efa5d688eb2d4615feb6585563972d5c28db35f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-05 12:20:42.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"57176d38c49b140d486dc256db70278a403dd825","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-05 12:21:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"87161e5788e49b5c590ff12f33436895b2d993fd","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-03-05 12:48:15.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"90e758c52c2b49fe0ebd5477328138990df2907c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-03-05 14:06:38.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"119a83aaf01a46df3bfc2ac59830e3f6562439f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 14:36:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bd1f0e38f884841b12c7c7e846cf005\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89bfdc16dea44134930dc0e57b836260 : SUCCESS in 1h 27m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/101934026cd84d03b2bd606ce0a09888 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21235ae26c4744ada6fa1f3ad4b1d2ae : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f25289946064aca9c112a6f80d0d5e1 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a42a05baf284069b4b6504e48a61204 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e674f2850c2a43e987ef7d77b22ec045 : SUCCESS in 4m 24s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/4f0f65a6c14e483e8bb866e75b28402a : SUCCESS in 8m 40s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/07971625e709436abe5bd232da1cd9ef : SUCCESS in 2h 15m 40s","accounts_in_message":[],"_revision_number":2},{"id":"263816a385f6902ead47511dc225af07eb9bbc12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-05 15:09:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ddf24257363d41afd2168b6357d007e9f54ae16e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-05 15:09:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a80d1751f35a9cc835e40655cc1af9c6caee6a3e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-03-05 18:05:03.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4cc80cbdc8bc7901e31ff5205ee10cce11610c6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 18:09:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e670f4f318a245f7878600a5cbcafd93\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6edaff2459fd40ce8945f070dc1d5e84 : SUCCESS in 1h 50m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6e0686e1c944359a6fc8e2183177f92 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c934a1c35574bd99f6768f9e97f9e3c : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb78168bdd2a477cbf7af2d0f02b289f : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41a233de5cfb4a1786e81db022d302ce : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/502c832dad1b4f3bb7aa7c26ae546824 : SUCCESS in 6m 32s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/1b58ec22444748a48c457faebabd53f8 : SUCCESS in 8m 41s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/a455ea5993f549aab16b64c02caeb042 : SUCCESS in 2h 52m 10s","accounts_in_message":[],"_revision_number":3},{"id":"d6afbc89442f61248770895c1607e8b35266e951","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-03-06 06:44:30.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b3542f580cfcc9822242d998c84b2ff31a64d29d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-07 12:50:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"22135abbb756baaba0fa1b47168d1db79f92223f","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-03-07 13:05:14.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"af0720e77a7fb2198467a5c444638f1ca62fa56f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-03-08 09:03:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"337638b867722d7bbef55f60f01c28f1623d6c02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-11 19:25:52.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c09a98370d80127cfb03f627b337ffbf58b61a69","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-11 19:27:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ee82f5074f1cc9bd5f4b2bad973778f7b58d65df","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-03-11 19:40:43.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4c269dae02cab68f408ff5b9fa8cb625325134f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 21:43:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e37da814a09342f5b8e6565aa19d9acf\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/282c6439838641e79bc2f1a6865e36e8 : SUCCESS in 1h 32m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d149da0562864c10ab99d6539dcd1431 : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/313d7028025043109015e5b2f0fc276b : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1987c18deb84fbaae57d16408957fb1 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22efad0fc6264b16b9f0f9ee9a1fc7a8 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/683085834103488a9a24d0178b0866bc : SUCCESS in 4m 02s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/d26a22c034e848c4b4e58ac5c17965f5 : SUCCESS in 10m 09s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/9fc295ee91914d4aa85886ae285f363f : SUCCESS in 2h 15m 25s","accounts_in_message":[],"_revision_number":4},{"id":"74ac19bfdc654ab97bf49155f4a5355f10f5ce39","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-04-04 14:23:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"85f0dc6f927f77c789cae02c8eca8a42614a9fb9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-04-04 14:34:34.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5351018b9ca91282b1826ef7b935fc8cbab24717","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-04-08 18:06:40.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"36e28af9c3810f9f387a97c33ba44640384ff4ff","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-04-08 18:07:29.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"40cf102941072201482ed5d46b5104c079f69c6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 20:38:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b942dcdfcbd4940aa9f51d68e72bcc6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/03e09eede36e42d986e7f14369b6c3f7 : SUCCESS in 1h 44m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91be5232b52440bbaa5187070e4abd3e : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33ac8fe71d564aa3a69fae90c06acbc4 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc206e0b98ec4183a9941a419f471f9b : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/63f2c5861e164632a869edc3e2886e37 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b40a84c62e6344dfadac11c28fe87ee8 : SUCCESS in 3m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74a7cdc3ce5441029f7c06a04aade2e6 : SUCCESS in 2m 42s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/0ef871f554ea4ee8980120a81922b014 : SUCCESS in 9m 00s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/4ba3d0010670477cb1c4309bb92d3e76 : SUCCESS in 2h 31m 43s","accounts_in_message":[],"_revision_number":5},{"id":"4b3c0e37f686053c5e16603147a872765b359120","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-04-09 05:04:34.000000000","message":"Patch Set 5: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5acc1e517bd9a672d9f8381d30171c5de01da3ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-04-09 06:01:17.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"3591009e0829dd23b11014495bde2adc3684be5a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-04-09 06:02:55.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ea9857d00edf8e5ea1d7f1a58d2929983eb095c3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-04-09 06:30:07.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f7812fab9fd898159ee691f38155d3f553e86c68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 08:32:33.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31c1575490494a248f59e07cc3b6ca45\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ebeae5af03a243058f87d66cec5f2636 : SUCCESS in 1h 01m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44875945ff7e4e5fb5911d734568b2c1 : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1296c5aed0614b8b94092f74fe67db1e : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d35632b9c24d411fa6684778ba10f8fa : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af0bd9b373684f5b9df5dfc5b11516b0 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c73e32edf7e4cf0914596f09df3dc57 : SUCCESS in 3m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f0e099c2e6d4ec18ac4bd47c0a869b0 : SUCCESS in 3m 56s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/f444ab9c5fe9468aa09acc0cdfa2895b : SUCCESS in 7m 04s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/aea94a6d6d9b41b98d1f69e1315220cc : SUCCESS in 2h 25m 59s","accounts_in_message":[],"_revision_number":6},{"id":"6b1f134259843f15cd1e76b47fc602550edd2d92","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-04-22 10:31:52.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"224074ae82aa38c12971cb83637075af8e5438c7","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-04-22 14:00:32.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"7e6ca72a07bfa3b7059ab09037538f3fc4d2cd53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 14:01:04.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"cb672a7c1f3d11d1dcdad37ba0771c59fd0f4a85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 15:46:09.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/990ab74993d24e22a0401309775840ea\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7cc05ed7d3840c5909adcf91f8d97ef : SUCCESS in 1h 40m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/554f51ef4d814aabb06550ac7f7823a7 : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bd82ad56a5f4519b8bd18546b44c210 : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec7bb98946424eb1a9d44402639e8a7d : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eefda11ac15431695598a2b4d144043 : SUCCESS in 3m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79a3ca408da04d4bbe6b0ed38af9be39 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":6},{"id":"e4f36d78f3511d319450dfd67cbef03588d75178","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 15:46:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"56d4f516024758f1d78aa0ca63faf0d0db3f687d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 15:47:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d09e121e54948979903f2df2d5c64cc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc2d171e15df4a499978ab176e20e5f4 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3b0d628361342c88318dc70230d1877 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c2b7b78d4ff55ea079ea6172241011edc68f5b23","revisions":{"520076019319da897de5cd4747291b2f1f9aa60e":{"kind":"REWORK","_number":1,"created":"2024-03-04 22:06:34.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/05/911005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/05/911005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/05/911005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/05/911005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/05/911005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/05/911005/1"}}},"commit":{"parents":[{"commit":"e9217809e76d60ea48d3853d9edd817ee077296f","subject":"Merge \"Bump hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/e9217809e76d60ea48d3853d9edd817ee077296f"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-03-04 21:56:55.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-03-04 22:06:30.000000000","tz":330},"subject":"Use normal credentials for legacy image update","message":"Use normal credentials for legacy image update\n\nWhen updating legacy images, we use the user\u0027s context and elevate\npriviledges. However, neither do we require admin priviledges for\nthe cinder API calls (in this case) nor the user\u0027s context credentials.\n\nThis patch removes the special case where we elevate priviledges as\nit wasn\u0027t doing anything rather avoiding us to use right credentials\nand failing to fetch volume in the right location because of wrong\ncredentials.\n\nNOTE: When using cinder as glance backend and we want to perform optimized\nvolume upload to image, one thing we should make sure is either using the\ncontext or the cinder credentials set in glance-api.conf file, it should match\nthe following details on the cinder side (if we are using internal context\nto create clone of image-volumes):\n\ncinder_store_user_name \u003d context.user_id \u003d cinder_internal_tenant_user_id\ncinder_store_project_name \u003d context.project_id \u003d cinder_internal_tenant_project_id\n\nThe cinder_internal_tenant_user_id and cinder_internal_tenant_project_id\nare set in the [DEFAULT] section of cinder.conf.\n\nThis issue was first discovered when testing the new location APIs[1] where\ntempest creates a volume with cinder\u0027s internal context and glance uses\nwrong credentials to access it and failing with 404 not found.\n\n[1] https://review.opendev.org/c/openstack/cinder/+/909847\n\nChange-Id: I4f27a9800f239da8dbf29f4c028678df1f867664\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/520076019319da897de5cd4747291b2f1f9aa60e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/520076019319da897de5cd4747291b2f1f9aa60e"}]},"branch":"refs/heads/master"},"82aee3b195bad7db3647a517bad84e994672aa56":{"kind":"REWORK","_number":2,"created":"2024-03-05 12:19:58.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/05/911005/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/05/911005/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/05/911005/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/05/911005/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/05/911005/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/05/911005/2"}}},"commit":{"parents":[{"commit":"e9217809e76d60ea48d3853d9edd817ee077296f","subject":"Merge \"Bump hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/e9217809e76d60ea48d3853d9edd817ee077296f"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-03-04 21:56:55.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-03-05 12:19:26.000000000","tz":330},"subject":"Use normal credentials for legacy image update","message":"Use normal credentials for legacy image update\n\nWhen updating legacy images, we use the user\u0027s context and elevate\npriviledges. However, neither do we require admin priviledges for\nthe cinder API calls (in this case) nor the user\u0027s context credentials.\n\nThis patch removes the special case where we elevate priviledges as\nit wasn\u0027t doing anything rather avoiding us to use right credentials\nand failing to fetch volume in the right location because of wrong\ncredentials.\n\nNOTE: When using cinder as glance backend and we want to perform optimized\nvolume upload to image, one thing we should make sure is either using the\ncontext or the cinder credentials set in glance-api.conf file, it should match\nthe following details on the cinder side (if we are using internal context\nto create clone of image-volumes):\n\ncinder_store_user_name \u003d context.user_id \u003d cinder_internal_tenant_user_id\ncinder_store_project_name \u003d context.project_id \u003d cinder_internal_tenant_project_id\n\nThe cinder_internal_tenant_user_id and cinder_internal_tenant_project_id\nare set in the [DEFAULT] section of cinder.conf.\n\nThis issue was first discovered when testing the new location APIs[1] where\ntempest creates a volume with cinder\u0027s internal context and glance uses\nwrong credentials to access it and failing with 404 not found.\n\n[1] https://review.opendev.org/c/openstack/cinder/+/909847\n\nCloses-Bug: #2056179\nChange-Id: I4f27a9800f239da8dbf29f4c028678df1f867664\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/82aee3b195bad7db3647a517bad84e994672aa56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/82aee3b195bad7db3647a517bad84e994672aa56"}]},"branch":"refs/heads/master"},"4de86d30a4a208e2fe25507b70e44867b2877564":{"kind":"REWORK","_number":3,"created":"2024-03-05 15:09:28.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/05/911005/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/05/911005/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/05/911005/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/05/911005/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/05/911005/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/05/911005/3"}}},"commit":{"parents":[{"commit":"e9217809e76d60ea48d3853d9edd817ee077296f","subject":"Merge \"Bump hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/e9217809e76d60ea48d3853d9edd817ee077296f"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-03-04 21:56:55.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-03-05 12:22:29.000000000","tz":330},"subject":"Use normal credentials for legacy image update","message":"Use normal credentials for legacy image update\n\nWhen updating legacy images, we use the user\u0027s context and elevate\npriviledges. However, neither do we require admin priviledges for\nthe cinder API calls (in this case) nor the user\u0027s context credentials.\n\nThis patch removes the special case where we elevate priviledges as\nit wasn\u0027t doing anything rather avoiding us to use right credentials\nand failing to fetch volume in the right location because of wrong\ncredentials.\n\nNOTE: When using cinder as glance backend and we want to perform optimized\nvolume upload to image, one thing we should make sure is either using the\ncontext or the cinder credentials set in glance-api.conf file, it should match\nthe following details on the cinder side (if we are using internal context\nto create clone of image-volumes):\n\ncinder_store_user_name \u003d context.user_id \u003d cinder_internal_tenant_user_id\ncinder_store_project_name \u003d context.project_id \u003d cinder_internal_tenant_project_id\n\nThe cinder_internal_tenant_user_id and cinder_internal_tenant_project_id\nare set in the [DEFAULT] section of cinder.conf.\n\nThis issue was first discovered when testing the new location APIs[1] where\ntempest creates a volume with cinder\u0027s internal context and glance uses\nwrong credentials to access it and failing with 404 not found.\n\n[1] https://review.opendev.org/c/openstack/cinder/+/909847\n\nCloses-Bug: #2056179\nChange-Id: I4f27a9800f239da8dbf29f4c028678df1f867664\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/4de86d30a4a208e2fe25507b70e44867b2877564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/4de86d30a4a208e2fe25507b70e44867b2877564"}]},"branch":"refs/heads/master"},"5062a9210ac90760062fe381a544d301bb5f15ca":{"kind":"REWORK","_number":4,"created":"2024-03-11 19:25:52.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/05/911005/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/05/911005/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/05/911005/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/05/911005/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/05/911005/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/05/911005/4"}}},"commit":{"parents":[{"commit":"e9217809e76d60ea48d3853d9edd817ee077296f","subject":"Merge \"Bump hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/e9217809e76d60ea48d3853d9edd817ee077296f"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-03-04 21:56:55.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-03-07 14:34:47.000000000","tz":330},"subject":"Use normal credentials for legacy image update","message":"Use normal credentials for legacy image update\n\nWhen updating legacy images, currenly we use the user\u0027s\ncontext and elevate priviledges. However, we do not\nrequire admin priviledges for the cinder API calls.\n\nThis patch removes the special case where we elevate\npriviledges as it wasn\u0027t doing anything rather avoiding\nus to use right credentials and failing to fetch volume\nin the right location because of wrong credentials.\n\nThe correct credentials are either the service ones\nset in glance-api.conf file or the user context\ncredentials, using which the Image-Volume was created.\n\nNOTE: When using cinder as glance backend and we want\nto perform optimized volume upload to image, one thing\nwe should make sure is either using the context or the\ncinder credentials set in glance-api.conf file, it should\nmatch the following details on the cinder side (if we are\nusing internal context to create clone of image-volumes):\n\ncinder_store_user_name \u003d context.user_id \u003d cinder_internal_tenant_user_id\ncinder_store_project_name \u003d context.project_id \u003d cinder_internal_tenant_project_id\n\nThe cinder_internal_tenant_user_id and\ncinder_internal_tenant_project_id are set in the\n[DEFAULT] section of cinder.conf.\n\nThis issue was first discovered when testing the new\nlocation APIs[1] where tempest creates a volume with\ncinder\u0027s internal context and glance uses wrong (user context)\ncredentials to access it and failing with 404 not found.\n\n[1] https://review.opendev.org/c/openstack/cinder/+/909847\n\nCloses-Bug: #2056179\nChange-Id: I4f27a9800f239da8dbf29f4c028678df1f867664\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/5062a9210ac90760062fe381a544d301bb5f15ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/5062a9210ac90760062fe381a544d301bb5f15ca"}]},"branch":"refs/heads/master"},"392217486ffc5383b5b12d7cc83a9554fb9bb2dd":{"kind":"REWORK","_number":5,"created":"2024-04-08 18:06:40.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/05/911005/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/05/911005/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/05/911005/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/05/911005/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/05/911005/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/05/911005/5"}}},"commit":{"parents":[{"commit":"e9217809e76d60ea48d3853d9edd817ee077296f","subject":"Merge \"Bump hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/e9217809e76d60ea48d3853d9edd817ee077296f"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-03-04 21:56:55.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-04-08 18:06:12.000000000","tz":330},"subject":"Use normal credentials for legacy image update","message":"Use normal credentials for legacy image update\n\nWhen updating legacy images, currenly we use the user\u0027s\ncontext and elevate priviledges. However, we do not\nrequire admin priviledges for the cinder API calls.\n\nThis patch removes the special case where we elevate\npriviledges as it wasn\u0027t doing anything rather avoiding\nus to use right credentials and failing to fetch volume\nin the right location because of wrong credentials.\n\nThe correct credentials are either the service ones\nset in glance-api.conf file or the user context\ncredentials, using which the Image-Volume was created.\n\nNOTE: When using cinder as glance backend and we want\nto perform optimized volume upload to image, one thing\nwe should make sure is either using the context or the\ncinder credentials set in glance-api.conf file, it should\nmatch the following details on the cinder side (if we are\nusing internal context to create clone of image-volumes):\n\ncinder_store_user_name \u003d context.user_id/cinder_internal_tenant_user_id\ncinder_store_project_name \u003d context.project_id/cinder_internal_tenant_project_id\n\nThe cinder_internal_tenant_user_id and\ncinder_internal_tenant_project_id are set in the\n[DEFAULT] section of cinder.conf.\n\nThis issue was first discovered when testing the new\nlocation APIs[1] where tempest creates a volume with\ncinder\u0027s internal context and glance uses wrong (user context)\ncredentials to access it and failing with 404 not found.\n\n[1] https://review.opendev.org/c/openstack/cinder/+/909847\n\nCloses-Bug: #2056179\nChange-Id: I4f27a9800f239da8dbf29f4c028678df1f867664\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/392217486ffc5383b5b12d7cc83a9554fb9bb2dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/392217486ffc5383b5b12d7cc83a9554fb9bb2dd"}]},"branch":"refs/heads/master"},"c2b7b78d4ff55ea079ea6172241011edc68f5b23":{"kind":"REWORK","_number":6,"created":"2024-04-09 06:01:17.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/05/911005/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/05/911005/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/05/911005/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/05/911005/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/05/911005/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/05/911005/6"}}},"commit":{"parents":[{"commit":"e9217809e76d60ea48d3853d9edd817ee077296f","subject":"Merge \"Bump hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/e9217809e76d60ea48d3853d9edd817ee077296f"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-03-04 21:56:55.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-04-08 18:23:24.000000000","tz":330},"subject":"Use normal credentials for legacy image update","message":"Use normal credentials for legacy image update\n\nWhen updating legacy images, currenly we use the user\u0027s\ncontext and elevate priviledges. However, we do not\nrequire admin priviledges for the cinder API calls.\n\nThis patch removes the special case where we elevate\npriviledges as it wasn\u0027t doing anything rather avoiding\nus to use right credentials and failing to fetch volume\nin the right location because of wrong credentials.\n\nThe correct credentials are either the service ones\nset in glance-api.conf file or the user context\ncredentials, using which the Image-Volume was created.\n\nNOTE: When using cinder as glance backend and we want\nto perform optimized volume upload to image, one thing\nwe should make sure is either using the context or the\ncinder credentials set in glance-api.conf file, it should\nmatch the following details on the cinder side (if we are\nusing internal context to create clone of image-volumes):\n\ncinder_store_user_name \u003d context.user_id/cinder_internal_tenant_user_id\ncinder_store_project_name \u003d context.project_id/cinder_internal_tenant_project_id\n\nThe cinder_internal_tenant_user_id and\ncinder_internal_tenant_project_id are set in the\n[DEFAULT] section of cinder.conf.\n\nThis issue was first discovered when testing the new\nlocation APIs[1] where tempest creates a volume with\ncinder\u0027s internal context and glance uses wrong (user context)\ncredentials to access it and failing with 404 not found.\n\n[1] https://review.opendev.org/c/openstack/cinder/+/909847\n\nCloses-Bug: #2056179\nChange-Id: I4f27a9800f239da8dbf29f4c028678df1f867664\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/c2b7b78d4ff55ea079ea6172241011edc68f5b23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/c2b7b78d4ff55ea079ea6172241011edc68f5b23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
