)]}'
{"id":"openstack%2Fglance_store~935033","triplet_id":"openstack%2Fglance_store~stable%2F2024.1~I852beab63512f2469f92181a2ba427cbaa172ec2","project":"openstack/glance_store","branch":"stable/2024.1","topic":"bug/2030825-stable/2024.2-stable/2024.1","attention_set":{},"removed_from_attention_set":{"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2024-11-14 09:11:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I852beab63512f2469f92181a2ba427cbaa172ec2","subject":"S3: add option to specify a custom CA cert bundle","status":"MERGED","created":"2024-11-14 04:25:48.000000000","updated":"2024-11-14 09:11:57.000000000","submitted":"2024-11-14 09:11:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935033-bug/2030825-stable/2024.2-stable/2024.1","cherry_pick_of_change":934719,"cherry_pick_of_patch_set":1,"meta_rev_id":"cba88de003d5bb79af393e2bd84eae596a911e18","_number":935033,"virtual_id_number":935033,"owner":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-14 09:11:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-11-14 06:17:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-11-14 07:39:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-14 04:25:48.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2024-11-14 06:17:45.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-11-14 07:32:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2824c14c22989db2e6a021d35679fb5583abecfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2024-11-14 04:25:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"09f631b23ef11ceed4c580fdffc362a23aa41502","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2024-11-14 04:26:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"f891764eacd0bdb85c2d47e1aebeec9e6b1dda12","tag":"autogenerated:gerrit:restore","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2024-11-14 04:30:10.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"c5d990871b182bc695a32142ce2f33e93c933f5f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2024-11-14 04:36:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"e7bcc72da9dba10dce73099d22a813dd4e754b64","tag":"autogenerated:gerrit:restore","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2024-11-14 04:39:46.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"4e8729e88abd872d99db96102b810cddefd91d00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2024-11-14 04:40:22.000000000","message":"Patch Set 2: Cherry Picked from branch stable/2024.2.","accounts_in_message":[],"_revision_number":2},{"id":"31898ea584fb7d42bc7bd8127023c81d2d9b1d3b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-14 06:17:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6a411b22936cf17dfeaac8ea70c99f71d9fa0230","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 07:32:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcbd82ccc5124aaf811b83137e315e23\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85055593351c42ab89017b0ccb152b8a : SUCCESS in 1h 09m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e295bd35a547467aa1ef2770954edab1 : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a65ea15ce8e46cab52c43dd4690f8a6 : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/39dc9fa0e7db4fdeb009a42606f6c582 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/916046cee81942529dc3d5d11adf1252 : SUCCESS in 2m 31s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/bc87b60e19f64688bf7aaadb9f2fdaa9 : SUCCESS in 7m 20s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/b8f5f99bd2b244bd9cdeefeb30771b59 : SUCCESS in 2h 51m 35s","accounts_in_message":[],"_revision_number":2},{"id":"a0b1c1265007a13c323f5376e989371a47da8003","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-14 07:39:56.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd9b740f95eb5cee0796f3b2db231acd9128ce67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 07:40:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a80e2c94ebe60dd6bba791a8217fb2f47ba9c3ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 09:11:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/738d9764a669431ab21de01ad001f77f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f9a29e55aae427fafdfcfa573906bbe : SUCCESS in 1h 30m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dc1b729d2694ff2943ad6d3629e46f1 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c3ed7d41c344760b720cda60692a2be : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/60d8b6c094044ac18cfd7dc77e75adb5 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/845d1099019a43cfa9bb4c6091445eb3 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":2},{"id":"2756cb5d23d2c3c6533214f0853b42fb1c3e62e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 09:11:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cba88de003d5bb79af393e2bd84eae596a911e18","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 09:11:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76f22d1e7d6b4444b92b0d03f505e255\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1186a035a1cb461084647363f4639fed : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f76d051190d24eb2cc5a1fc4e97eacebf53a494b","revisions":{"1ed3cb31d272ac0ebfea44c2b676c5e0565d4219":{"kind":"REWORK","_number":1,"created":"2024-11-14 04:25:48.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/33/935033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/33/935033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/33/935033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/33/935033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/33/935033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/33/935033/1"}}},"commit":{"parents":[{"commit":"7a7e1f566f123847eb82b614deed41da31ab5504","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/7a7e1f566f123847eb82b614deed41da31ab5504"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-09-06 20:08:55.000000000","tz":120},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2024-11-14 04:25:48.000000000","tz":0},"subject":"S3: add option to specify a custom CA cert bundle","message":"S3: add option to specify a custom CA cert bundle\n\nUsers can:\n- set this option to the path to their CA cert bundle;\n- use the default value to use the default CA cert bundle used by\n  botocore.\n\nWe do not allow users to disable SSL certificates verification.\n\nCloses-Bug: #2030825\n\nChange-Id: I852beab63512f2469f92181a2ba427cbaa172ec2\n(cherry picked from commit f9db33edc78411a0d7cf82a19b01da24ad7304fd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/1ed3cb31d272ac0ebfea44c2b676c5e0565d4219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/1ed3cb31d272ac0ebfea44c2b676c5e0565d4219"}]},"branch":"refs/heads/stable/2024.1"},"f76d051190d24eb2cc5a1fc4e97eacebf53a494b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-11-14 04:40:22.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/33/935033/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/33/935033/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/33/935033/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/33/935033/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/33/935033/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/33/935033/2"}}},"commit":{"parents":[{"commit":"7a7e1f566f123847eb82b614deed41da31ab5504","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/7a7e1f566f123847eb82b614deed41da31ab5504"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-09-06 20:08:55.000000000","tz":120},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2024-11-14 04:40:22.000000000","tz":0},"subject":"S3: add option to specify a custom CA cert bundle","message":"S3: add option to specify a custom CA cert bundle\n\nUsers can:\n- set this option to the path to their CA cert bundle;\n- use the default value to use the default CA cert bundle used by\n  botocore.\n\nWe do not allow users to disable SSL certificates verification.\n\nCloses-Bug: #2030825\n\nChange-Id: I852beab63512f2469f92181a2ba427cbaa172ec2\n(cherry picked from commit f9db33edc78411a0d7cf82a19b01da24ad7304fd)\n(cherry picked from commit aeac1d7a0ac6dda16af9fe7312788a89255839ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/f76d051190d24eb2cc5a1fc4e97eacebf53a494b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/f76d051190d24eb2cc5a1fc4e97eacebf53a494b"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
