)]}'
{"id":"openstack%2Fglance_store~935503","triplet_id":"openstack%2Fglance_store~stable%2F2023.1~I852beab63512f2469f92181a2ba427cbaa172ec2","project":"openstack/glance_store","branch":"stable/2023.1","topic":"bug/2030825-stable/2024.2-stable/2024.1-stable/2023.2-stable/2023.1","attention_set":{},"removed_from_attention_set":{"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2024-11-29 13:22:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I852beab63512f2469f92181a2ba427cbaa172ec2","subject":"S3: add option to specify a custom CA cert bundle","status":"ABANDONED","created":"2024-11-18 11:44:04.000000000","updated":"2024-11-29 13:22:09.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":935264,"cherry_pick_of_patch_set":1,"meta_rev_id":"588e05f97c57a21f64da910665e89d042c9ac27e","_number":935503,"virtual_id_number":935503,"owner":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-11-18 12:09:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-18 11:44:04.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2024-11-18 12:09:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-22 05:37:15.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"}],"messages":[{"id":"57c193a85c325a486fdbe6b00a29734a6fd6ce2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2024-11-18 11:44:04.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"866979628b8d523b8aa1a510fd8a0945ad1129aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 12:09:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2ca5d037bad44ebaa7a3cb20c41a016\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b33b8e24620464fa08cb9751e6747bf : FAILURE in 19m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66b2f2f8276b4625bb744be0f58d3b19 : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/221e24a2a8b549d586586c749b87b9ae : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22883042e4f34461a0cee2c0ef8fe028 : SUCCESS in 3m 46s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/a38d0242f5da42bd87b993f5307bd980 : SUCCESS in 9m 03s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/1a97383adc60467184debe6e984970b8 : FAILURE in 11m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3e729654784c782198890ee51609ecba02961f9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-22 05:37:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"588e05f97c57a21f64da910665e89d042c9ac27e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-29 13:22:09.000000000","message":"Abandoned\n\nstable/2023.1 branch of this repo is about to be deleted.       \nTo be able to do that, all open patches need to be abandoned.\nPlease cherry pick the patch to unmaintained/2023.1 if you want to \nfurther work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da39b69db20da30f00d903764789edd3067c5c65","revisions":{"da39b69db20da30f00d903764789edd3067c5c65":{"kind":"REWORK","_number":1,"created":"2024-11-18 11:44:04.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/03/935503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/03/935503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/03/935503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/03/935503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/03/935503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/03/935503/1"}}},"commit":{"parents":[{"commit":"97b6f4c030189a80df0b525420a3760ae3a81941","subject":"Merge \"Increase timeout of glance_store-src-ceph-tempest\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/97b6f4c030189a80df0b525420a3760ae3a81941"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-09-06 20:08:55.000000000","tz":120},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2024-11-18 11:44:04.000000000","tz":0},"subject":"S3: add option to specify a custom CA cert bundle","message":"S3: add option to specify a custom CA cert bundle\n\nUsers can:\n- set this option to the path to their CA cert bundle;\n- use the default value to use the default CA cert bundle used by\n  botocore.\n\nWe do not allow users to disable SSL certificates verification.\n\nCloses-Bug: #2030825\n\nChange-Id: I852beab63512f2469f92181a2ba427cbaa172ec2\n(cherry picked from commit f9db33edc78411a0d7cf82a19b01da24ad7304fd)\n(cherry picked from commit aeac1d7a0ac6dda16af9fe7312788a89255839ce)\n(cherry picked from commit f76d051190d24eb2cc5a1fc4e97eacebf53a494b)\n(cherry picked from commit ac477fb66776214e20d5e85cb18ea5a548aa41cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/da39b69db20da30f00d903764789edd3067c5c65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/da39b69db20da30f00d903764789edd3067c5c65"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
