)]}'
{"id":"openstack%2Fglance_store~953796","triplet_id":"openstack%2Fglance_store~master~I7ff28b85987a78e9d2b31716b7b9e32c03c3858d","project":"openstack/glance_store","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-07-01 15:29:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7ff28b85987a78e9d2b31716b7b9e32c03c3858d","subject":"Replace os-client-config","status":"MERGED","created":"2025-06-30 15:53:53.000000000","updated":"2025-07-01 15:30:15.000000000","submitted":"2025-07-01 15:29:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953796","meta_rev_id":"a50cc2abe0e8a8dfeb671e7c894317652a3c687f","_number":953796,"virtual_id_number":953796,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-01 15:29:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-07-01 13:41:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2025-06-30 16:36:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2025-07-01 13:41:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-30 16:14:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-30 16:36:30.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-07-01 13:41:21.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"5c2c8ec0f25bc125a24569ec9b170d99a2a0c6c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-30 15:53:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27c5426dcc134f519c22240f8d879ef81632151d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-30 15:56:30.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1faf30ef0f7b294fa64f0d7d28a2042feb22b0f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-30 15:57:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"58a98f6e709956a9411fbc9e6a8fd4c3d5ddda99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-30 16:04:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"058106990cf590e315e49b6567f595356be84ff3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 16:14:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6920e2c1cc3e4cd38581c79a331a34c8\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8d5f4889291f4bab9ad0d32df1e9849e : FAILURE in 2m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6eb843fc141240fd9ddba26f8fa77867 : FAILURE in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4decb94443142468bee1abcd0906cb5 : FAILURE in 2m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92027e3091c64b15a48adf6215872e69 : FAILURE in 2m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/49c9782b56c84c9d9f51eea5000d247a : FAILURE in 2m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43d4f11413ce4531853b0eb1e496338a : FAILURE in 3m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bdadc5669a0436598c0e750010d1f3d : FAILURE in 2m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c9d48274ac54772a401780c11255ec4 : SUCCESS in 1m 44s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/e3f5566a146643979932273dbf64591a : FAILURE in 4m 50s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/935ad6e9d1d9479a91999acbc4b475e2 : FAILURE in 9m 05s","accounts_in_message":[],"_revision_number":4},{"id":"3a1eede61db81c0be8e04a7dd3994cd96aa5a7ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-30 16:15:13.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a5a61b5cb202748fbf3f5a82310efe84693bed05","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-06-30 16:36:30.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"bd7a3420f024dea6d4df4ef35d6b0e878194cdba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 18:49:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b3c21c5590d4318b3d40ffcfddb489a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8c0d04baec2c43179b6b0c84edf14b06 : SUCCESS in 1m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e4c955599524e919563633b7598bd5d : SUCCESS in 1h 42m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dd551fa3f86452fa211251842d6a0c3 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12218c503c444a229197dae7331a72cd : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/322fce32cc914a2f840777ea92b6603e : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/78939c0af79b4fb19c90c59322e53e9d : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0384e672e164470fb670d9c6d138465b : SUCCESS in 2m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6c02d57a3ae483e995cf342d794c0cd : SUCCESS in 2m 53s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/c6a9944ec999486a827877c5ccfa1007 : SUCCESS in 9m 11s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/c7da5b40fb6d41fb912d0f0a61270736 : SUCCESS in 2h 33m 13s","accounts_in_message":[],"_revision_number":5},{"id":"742f59e7f3aa367a819839f0fd3a71f36e795e67","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-01 13:41:21.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"cb2bf1ee31830efda2b31b6120b4b44b6e41b728","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 13:41:41.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c8e28c874b51724c3382f74652715b243912ae5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 15:29:17.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8052b01b7af41ff83b649605218e207\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/299912c1c6f64eaca47020e45e92a12c : SUCCESS in 2m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/de16f4a6a7384533aa9e25464aa5ae33 : SUCCESS in 1h 43m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4bb468c160c430f8d67e566c85d2ac3 : SUCCESS in 2m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43703b053eb847658719fa620cbed10b : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2ade5b057f4e4bcdab9b9dd8e207de0b : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68e522ce9b3749debb76c1225eaa628b : SUCCESS in 3m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f34f174b9517485fbfd54fcd7f0c7962 : SUCCESS in 2m 09s","accounts_in_message":[],"_revision_number":5},{"id":"ea598ceb51c6cee632b69918a548f0a582b5abf5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 15:29:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"a50cc2abe0e8a8dfeb671e7c894317652a3c687f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 15:30:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b26facab6af42e8a734ad3bbdb65564\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6851cce960224086ba0cc098be825db9 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/698d39bb88ec4d608a32a7fe1b55a36c : SUCCESS in 50s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a0997342aa79357ef210b6328994ee6344a1adf9","revisions":{"641be159d21e4f66c0dc92779ccc228ea1c9acad":{"kind":"REWORK","_number":1,"created":"2025-06-30 15:53:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/953796/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/96/953796/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/953796/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/953796/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/953796/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/96/953796/1"}}},"commit":{"parents":[{"commit":"aeb3fe04a4e4c73a7a384faea76fee6b2d31d5e0","subject":"Merge \"Drop explicit dependency on python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/aeb3fe04a4e4c73a7a384faea76fee6b2d31d5e0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-30 15:51:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-30 15:53:25.000000000","tz":540},"subject":"Replace os-client-config","message":"Replace os-client-config\n\nIt was deprecated[1] after the code was merged into openstacksdk[2].\n\n[1] https://review.opendev.org/c/openstack/os-client-config/+/549307\n[2] https://review.opendev.org/c/openstack/openstacksdk/+/518128\n\nChange-Id: I7ff28b85987a78e9d2b31716b7b9e32c03c3858d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/641be159d21e4f66c0dc92779ccc228ea1c9acad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/641be159d21e4f66c0dc92779ccc228ea1c9acad"}]},"branch":"refs/heads/master"},"afb812e46496a43d3d6c576ed7e5694045825b65":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-06-30 15:56:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/953796/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/96/953796/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/953796/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/953796/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/953796/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/96/953796/2"}}},"commit":{"parents":[{"commit":"aeb3fe04a4e4c73a7a384faea76fee6b2d31d5e0","subject":"Merge \"Drop explicit dependency on python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/aeb3fe04a4e4c73a7a384faea76fee6b2d31d5e0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-30 15:51:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-30 15:56:21.000000000","tz":540},"subject":"Replace os-client-config","message":"Replace os-client-config\n\nIt was deprecated[1] after the code was merged into openstacksdk[2].\n\n[1] https://review.opendev.org/c/openstack/os-client-config/+/549307\n[2] https://review.opendev.org/c/openstack/openstacksdk/+/518128\n\nChange-Id: I7ff28b85987a78e9d2b31716b7b9e32c03c3858d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/afb812e46496a43d3d6c576ed7e5694045825b65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/afb812e46496a43d3d6c576ed7e5694045825b65"}]},"branch":"refs/heads/master"},"0e1fac0f3f62296a01b6284390c3b94a059374cd":{"kind":"REWORK","_number":3,"created":"2025-06-30 15:57:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/953796/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/96/953796/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/953796/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/953796/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/953796/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/96/953796/3"}}},"commit":{"parents":[{"commit":"aeb3fe04a4e4c73a7a384faea76fee6b2d31d5e0","subject":"Merge \"Drop explicit dependency on python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/aeb3fe04a4e4c73a7a384faea76fee6b2d31d5e0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-30 15:51:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-30 15:57:38.000000000","tz":540},"subject":"Replace os-client-config","message":"Replace os-client-config\n\nIt was deprecated[1] after the code was merged into openstacksdk[2].\n\n[1] https://review.opendev.org/c/openstack/os-client-config/+/549307\n[2] https://review.opendev.org/c/openstack/openstacksdk/+/518128\n\nChange-Id: I7ff28b85987a78e9d2b31716b7b9e32c03c3858d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0e1fac0f3f62296a01b6284390c3b94a059374cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0e1fac0f3f62296a01b6284390c3b94a059374cd"}]},"branch":"refs/heads/master"},"15b0e13144b04ea8672f1b7b0a60dd63bec0e0b2":{"kind":"REWORK","_number":4,"created":"2025-06-30 16:04:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/953796/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/96/953796/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/953796/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/953796/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/953796/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/96/953796/4"}}},"commit":{"parents":[{"commit":"aeb3fe04a4e4c73a7a384faea76fee6b2d31d5e0","subject":"Merge \"Drop explicit dependency on python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/aeb3fe04a4e4c73a7a384faea76fee6b2d31d5e0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-30 15:51:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-30 16:04:39.000000000","tz":540},"subject":"Replace os-client-config","message":"Replace os-client-config\n\nIt was deprecated[1] after the code was merged into openstacksdk[2].\n\n[1] https://review.opendev.org/c/openstack/os-client-config/+/549307\n[2] https://review.opendev.org/c/openstack/openstacksdk/+/518128\n\nChange-Id: I7ff28b85987a78e9d2b31716b7b9e32c03c3858d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/15b0e13144b04ea8672f1b7b0a60dd63bec0e0b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/15b0e13144b04ea8672f1b7b0a60dd63bec0e0b2"}]},"branch":"refs/heads/master"},"a0997342aa79357ef210b6328994ee6344a1adf9":{"kind":"REWORK","_number":5,"created":"2025-06-30 16:15:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/953796/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/96/953796/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/953796/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/953796/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/953796/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/96/953796/5"}}},"commit":{"parents":[{"commit":"aeb3fe04a4e4c73a7a384faea76fee6b2d31d5e0","subject":"Merge \"Drop explicit dependency on python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/aeb3fe04a4e4c73a7a384faea76fee6b2d31d5e0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-30 15:51:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-30 16:15:01.000000000","tz":540},"subject":"Replace os-client-config","message":"Replace os-client-config\n\nIt was deprecated[1] after the code was merged into openstacksdk[2].\n\n[1] https://review.opendev.org/c/openstack/os-client-config/+/549307\n[2] https://review.opendev.org/c/openstack/openstacksdk/+/518128\n\nChange-Id: I7ff28b85987a78e9d2b31716b7b9e32c03c3858d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a0997342aa79357ef210b6328994ee6344a1adf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a0997342aa79357ef210b6328994ee6344a1adf9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
