)]}'
{"id":"openstack%2Fglance_store~961870","triplet_id":"openstack%2Fglance_store~stable%2F2024.2~I436d56d0489b081822ed55108e10d43cc38512bb","project":"openstack/glance_store","branch":"stable/2024.2","topic":"stable/2024.2-bug/2043740","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-05-07 13:01:08.000000000","reason":"Change was abandoned"},"8108":{"account":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"last_update":"2026-05-07 13:01:08.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I436d56d0489b081822ed55108e10d43cc38512bb","subject":"Support multipart upload when image_size \u003d\u003d 0","status":"ABANDONED","created":"2025-09-20 16:42:17.000000000","updated":"2026-05-07 13:01:08.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"551b259f2404b1f4ac1b306c854b9f659a603ae2","_number":961870,"virtual_id_number":961870,"owner":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-23 20:55:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-20 16:42:17.000000000","updated_by":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-09-20 19:24:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-24 18:23:30.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-10-23 15:37:03.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2026-05-07 13:01:08.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"0950e8043f556226bcf8afe49ab15919dc9ea23e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2025-09-20 16:42:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"195110be73731bb6f9edbfd262d066d13f92477d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-20 19:24:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf3ba66665c744fe9e9f4117a178252e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/925781a7382a4ab4bd785a46be3ed2f3 : SUCCESS in 1h 19m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b5e4b1d0d584fdba3d2c7de6e6bc252 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35e97118beea44578900641544c0b548 : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a685a19b75f449359fbe1cae53656c27 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a9a34ed15504d9793ae7fd00df60546 : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/292fe68e7a3f4fc19f042e5f29be574e : SUCCESS in 2m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/788567d3cb31484a82c2244808256c38 : SUCCESS in 3m 13s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/5142d9b6e4bf4043a2a95925f92aecdc : FAILURE in 4m 31s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/3a93c909f43845178ac23ff49217580a : SUCCESS in 2h 37m 56s","accounts_in_message":[],"_revision_number":1},{"id":"80a09ff21eba3bf5f3675a078a64b94bd432c117","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-09-24 18:23:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a446d26fe7c7502815ec2adcc7adaad42b8715b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-10-16 19:59:45.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bb21bb00142c0fb929817e5ed457a464bf32959f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 21:42:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/298722101c9b413a9ccd780ce57c7aad\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/155e4ddcb33641c3a76c59d88a1a1c88 : SUCCESS in 1h 34m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47786abd6e8345018c095c4fc85e397e : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e119a5d64e6492c8fba5e9b49af1104 : SUCCESS in 3m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f30f5581146463aaa86a89d8cce16ad : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d09deb87d8545abb6f2542a54249f7d : SUCCESS in 2m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2b92c27ba0e4bc3b007142d91681d77 : SUCCESS in 1m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f9a9451e6cb42cf9680077b4385c351 : SUCCESS in 2m 28s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/a81ab6fcc01a46bc8ef4c04a753884bd : FAILURE in 5m 50s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/a381d2eaaf3f4fc987c9abbbbef5461a : SUCCESS in 1h 41m 52s","accounts_in_message":[],"_revision_number":2},{"id":"f5f854f0e64aa12be950de8137469089cea75e12","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-10-16 23:18:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cc6899c3029a67612aa295ad054b575e39dde7fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 02:05:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c7f0ce3cb3c4ece85d320e58f4b97e6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/19ac19d033774c8b9d60f125f7950ced : SUCCESS in 1h 42m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fb8ede1150043d7900a3a2c8c66833d : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82ca6c7eea5448609fdca4fb50fc21ad : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9646fe1b66a747798488d8dc15f8e98a : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b80bf188dfb64f71878f5e10c8e730d5 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9e1c3d27e5b45988d51366e2b09d364 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f8ee75c070b4323a73690d0efb45bf9 : SUCCESS in 1m 44s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/91fd0f107659404aa9f3447fee3951a1 : FAILURE in 7m 10s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/7d50886e26e14d52bd95db55c4d49c47 : SUCCESS in 2h 42m 10s","accounts_in_message":[],"_revision_number":2},{"id":"4426e456cf6942efa7d74e37b07f4a7c38466fdd","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-23 15:37:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4ae86d055e247f57f0515e038a16b24121250447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 18:45:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0a6a91223fb4c519ecdde77c1fa9ebb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a7165a0db3c4480be2070cfeea97ade : SUCCESS in 1h 09m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f5759d2f0c74a50b0e13431f976ae9f : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/553f1f1b6f5349b39f0fe427eba225a9 : SUCCESS in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ee6fe3bb47442e1834f0f4f847d92af : SUCCESS in 1m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/544cb562051e4eb3842e451f1a400b07 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd60478832104e059d6b3d85f8033fbc : SUCCESS in 2m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2433f57400a547a4956a697f7c23f364 : SUCCESS in 1m 44s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/15fa157c73ec4efca138eb933e2d8ca6 : FAILURE in 4m 15s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/b47b7a7b912d4053a2edc4d0521fda98 : TIMED_OUT in 3h 03m 39s","accounts_in_message":[],"_revision_number":2},{"id":"daafbdb562a60fa7a3cea2cf07d63a56aa719579","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-23 18:48:05.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b2e15a5fdffe6b7f2e1bd6c5160ec404fb4bc85d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 20:55:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72efdfb9348d41f59706e2b42e4bb0ff\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4fcfb19f97d74d9ca7fc8e7c638e582a : SUCCESS in 48m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5cd05286eb24ab29bd844b81982fc58 : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50cfba9f4f9f4432b78d41fc398d4933 : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6439b00834b649acbc0b50e3b2227211 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d70c61b1d6c34916b50200669dc866f3 : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc134437aefe4311ac73f5eb1a6584dc : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b32b0263b5a42f794fbf56e077d4e4d : SUCCESS in 2m 05s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/9736bd34a5b64d32bb4c97b1408745b4 : SUCCESS in 9m 42s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/98759873a1af4a89ad967c7f3d322c97 : SUCCESS in 2h 06m 55s","accounts_in_message":[],"_revision_number":3},{"id":"4e5a541cdbeb2379cbb5323623bb475d7585871f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-23 21:04:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"551b259f2404b1f4ac1b306c854b9f659a603ae2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-05-07 13:01:08.000000000","message":"Abandoned\n\nstable/2024.2 branch of openstack/glance_store transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7f90a22736ab46fc37a9aaa77247486ba1d44577","revisions":{"d9fad7329d1080ed1b4a68e6bcfcccc808917477":{"kind":"REWORK","_number":1,"created":"2025-09-20 16:42:17.000000000","uploader":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"ref":"refs/changes/70/961870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/70/961870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/70/961870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/70/961870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/70/961870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/70/961870/1"}}},"commit":{"parents":[{"commit":"6182b2b2fe640df4c763ae8fc6c04571374479a1","subject":"Add interface to get store weight from memory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6182b2b2fe640df4c763ae8fc6c04571374479a1"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-04-04 20:16:47.000000000","tz":120},"committer":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2025-09-20 16:42:05.000000000","tz":540},"subject":"Support multipart upload when image_size \u003d\u003d 0","message":"Support multipart upload when image_size \u003d\u003d 0\n\nWhen doing an image upload of a volume from cinder-volume to glance with\nan s3 backend, image_size in this context is zero. As a result the check\nthat image_size is larger than s3_store_large_object_size is always\nnegative and glance_store never uses multipart upload.\n\nAs a workaround, allow setting s3_store_large_object_size to 0, which\nforces glance_store to use multipart upload.\n\nRelated-Bug: #2043740\nChange-Id: I436d56d0489b081822ed55108e10d43cc38512bb\n(cherry picked from commit e63e0e4008971ebf8426dabd3abb4290a3d737f2)\nSigned-off-by: Nobuto Murata \u003cnobuto.murata@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d9fad7329d1080ed1b4a68e6bcfcccc808917477"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d9fad7329d1080ed1b4a68e6bcfcccc808917477"}]},"branch":"refs/heads/stable/2024.2"},"14dea6d4c411c4d79dc6ce4915625fc54055412d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-10-16 19:59:45.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/70/961870/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/70/961870/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/70/961870/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/70/961870/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/70/961870/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/70/961870/2"}}},"commit":{"parents":[{"commit":"6182b2b2fe640df4c763ae8fc6c04571374479a1","subject":"Add interface to get store weight from memory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6182b2b2fe640df4c763ae8fc6c04571374479a1"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-04-04 20:16:47.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-10-16 19:58:53.000000000","tz":120},"subject":"Support multipart upload when image_size \u003d\u003d 0","message":"Support multipart upload when image_size \u003d\u003d 0\n\nWhen doing an image upload of a volume from cinder-volume to glance with\nan s3 backend, image_size in this context is zero. As a result the check\nthat image_size is larger than s3_store_large_object_size is always\nnegative and glance_store never uses multipart upload.\n\nAs a workaround, allow setting s3_store_large_object_size to 0, which\nforces glance_store to use multipart upload.\n\nRelated-Bug: #2043740\nChange-Id: I436d56d0489b081822ed55108e10d43cc38512bb\n(cherry picked from commit e63e0e4008971ebf8426dabd3abb4290a3d737f2)\nSigned-off-by: Nobuto Murata \u003cnobuto.murata@canonical.com\u003e\n(cherry picked from commit 5db60a4a979662d2a09d937cf460f8d72f8403dd)\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/14dea6d4c411c4d79dc6ce4915625fc54055412d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/14dea6d4c411c4d79dc6ce4915625fc54055412d"}]},"branch":"refs/heads/stable/2024.2"},"7f90a22736ab46fc37a9aaa77247486ba1d44577":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-10-23 18:48:05.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/70/961870/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/70/961870/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/70/961870/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/70/961870/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/70/961870/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/70/961870/3"}}},"commit":{"parents":[{"commit":"6182b2b2fe640df4c763ae8fc6c04571374479a1","subject":"Add interface to get store weight from memory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6182b2b2fe640df4c763ae8fc6c04571374479a1"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-04-04 20:16:47.000000000","tz":120},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2025-10-23 18:48:05.000000000","tz":0},"subject":"Support multipart upload when image_size \u003d\u003d 0","message":"Support multipart upload when image_size \u003d\u003d 0\n\nWhen doing an image upload of a volume from cinder-volume to glance with\nan s3 backend, image_size in this context is zero. As a result the check\nthat image_size is larger than s3_store_large_object_size is always\nnegative and glance_store never uses multipart upload.\n\nAs a workaround, allow setting s3_store_large_object_size to 0, which\nforces glance_store to use multipart upload.\n\nDepends-on: https://review.opendev.org/c/zuul/zuul-jobs/+/964724\n\nRelated-Bug: #2043740\nChange-Id: I436d56d0489b081822ed55108e10d43cc38512bb\n(cherry picked from commit e63e0e4008971ebf8426dabd3abb4290a3d737f2)\nSigned-off-by: Nobuto Murata \u003cnobuto.murata@canonical.com\u003e\n(cherry picked from commit 5db60a4a979662d2a09d937cf460f8d72f8403dd)\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/7f90a22736ab46fc37a9aaa77247486ba1d44577"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/7f90a22736ab46fc37a9aaa77247486ba1d44577"}]},"branch":"refs/heads/stable/2024.2","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
