)]}'
{"id":"openstack%2Fglance_store~966449","triplet_id":"openstack%2Fglance_store~stable%2F2025.1~I5a4bad40ea03b0d746aaad392781f1b0fe78c8bc","project":"openstack/glance_store","branch":"stable/2025.1","topic":"fix-multiattach-handling","attention_set":{},"removed_from_attention_set":{"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2026-07-10 11:46:18.000000000","reason":"Change was abandoned"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-07-10 11:46:18.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5a4bad40ea03b0d746aaad392781f1b0fe78c8bc","subject":"Cinder: Standardize volume GET calls","status":"ABANDONED","created":"2025-11-07 15:24:27.000000000","updated":"2026-07-10 11:46:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6c456bfbe2fc15f049541f2f630de117336a73b9","_number":966449,"virtual_id_number":966449,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-11-07 18:44:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-07 15:24:27.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-11-07 18:44:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-06 14:33:27.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"}],"messages":[{"id":"31aeb7deea877c265908f7f75610e38bc2dc2c63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2025-11-07 15:24:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6b063d4ec2fa76c8d85191c792475f79988f9de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 18:44:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8bc3dbb0733c4dffbc7e045df6e92097\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7eb02186862b482c80c1c640dd0aa10d : SUCCESS in 1h 24m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b15981154b1541219d37753482486134 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b9d33772e244672931f005b71d2fc72 : SUCCESS in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2cc3c8ae6d0043d790f8f1ee93776f3c : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d01d4c8a6f74de08fb550fe4db68e58 : SUCCESS in 3m 58s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/82301b7de3a2484db0b2f20caf015916 : FAILURE in 9m 14s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/0d535722960f445caaf9ebc431d1c9a2 : TIMED_OUT in 3h 06m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7ebe71aa1b5a4e78dda020f5455438ed15c1d0fb","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-05-06 14:33:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"75df277deeee1c4a901b991abdd962c075c07fa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 17:44:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dba2c524fd4a4b449a4d56c2636a45e2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a837ffbadbd44df88583bb54f1cafb2 : SUCCESS in 57m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44ad6237d7b74e0683b5ed833c113a49 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d6a8d3b534843edbeeef4e4c93166d7 : SUCCESS in 4m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd85a525a012443582b1e8f78c3a752f : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51a96dff72db47fc9fe2386f4cb3eb6f : SUCCESS in 3m 25s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/c63f72cd7a3542ec9afc26303b4c1ff6 : FAILURE in 6m 16s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/1acdc82415be45c2b439715a6cc731b1 : TIMED_OUT in 3h 06m 05s","accounts_in_message":[],"_revision_number":1},{"id":"4c8fc70faa7fb173272a89b5fdf26a1dd6fdaa65","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-06-01 13:42:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"467a7cdb9958c6a899f4a804be1519315704b44f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 15:29:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3033368deda545f885e3ed041ba9db2c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78807810fc2448eb8651c42c251204cc : SUCCESS in 1h 05m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f100092973d4a20a21570755caeb571 : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b91b205925f4b6d8c011d3503fdd300 : SUCCESS in 3m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ddba4db359d742158c22835622544bce : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af24533d850b41d09b31b13339330d3a : SUCCESS in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37b73a710858402bbd85afe80fe69b0d : SUCCESS in 3m 03s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/92a0b34a38d049dbb3d199a06a4b4590 : FAILURE in 9m 49s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/cfd298f3b5e94c549f97c9a04aca8fbe : SUCCESS in 1h 43m 02s","accounts_in_message":[],"_revision_number":1},{"id":"6c456bfbe2fc15f049541f2f630de117336a73b9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2026-07-10 11:46:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d966bd20a436b0ba200e3e67bd2a51fe1a0679aa","revisions":{"d966bd20a436b0ba200e3e67bd2a51fe1a0679aa":{"kind":"REWORK","_number":1,"created":"2025-11-07 15:24:27.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/49/966449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/49/966449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/966449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/966449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/966449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/49/966449/1"}}},"commit":{"parents":[{"commit":"5db60a4a979662d2a09d937cf460f8d72f8403dd","subject":"Support multipart upload when image_size \u003d\u003d 0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/5db60a4a979662d2a09d937cf460f8d72f8403dd"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-09-26 19:42:51.000000000","tz":0},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2025-11-07 15:21:35.000000000","tz":60},"subject":"Cinder: Standardize volume GET calls","message":"Cinder: Standardize volume GET calls\n\nThe logic to fetch a volume details is scattered all over which\nmakes it hard to read as well as maintain.\nThere are currently two ways we fetch volume details:\n1. client.volumes.get\n2. volume.manager.get\n\nThe first one being the correct way maintaining abstraction and\nusing the cinderclient for API calls.\nThe problem with current code is:\n1. Hard to read and maintain\n2. Any change in volume GET logic will require changes across\nwhole store\n3. Not handling exceptions in a common place\n4. vulnerable to bugs as issue in one place cannot be detected\neasily by other code paths and needs to be isolately tested\n\nThis patch aims to standardize the volume GET call so we can\navoid the above stated issues.\nThis patch has been tested with the following operations massaging\nevery place modified in this patch:\n1. Create image - _wait_volume_status\n2. Create \u003e 1 GB image\n2.1 offline - _call_offline_extend\n2.2 online - _call_online_extend\n3. upload volume to image - is_image_associated_with_store\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/962445\n\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nChange-Id: I5a4bad40ea03b0d746aaad392781f1b0fe78c8bc\n(cherry picked from commit 3d55084cde7f30aba2ec41c003242d2762cf4413)\nSigned-off-by: Bence Romsics \u003cbence.romsics@gmail.com\u003e\n(cherry picked from commit 6943ba10313128e1e83c804c273b2cf8d220ffaa)\nConflicts:\n    glance_store/tests/unit/cinder/test_cinder_base.py\n        conflict with 3d57de1f5a9f83cf5e4d0105828959341557a4d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d966bd20a436b0ba200e3e67bd2a51fe1a0679aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d966bd20a436b0ba200e3e67bd2a51fe1a0679aa"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
