)]}'
{"id":"openstack%2Fglance_store~974096","triplet_id":"openstack%2Fglance_store~master~I57e70989f6bcab586a28ec150ac6b6c5d8a86918","project":"openstack/glance_store","branch":"master","attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-02-17 16:53:28.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-30 08:42:08.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I57e70989f6bcab586a28ec150ac6b6c5d8a86918","subject":"Migrate setup configuration to pyproject.toml","status":"NEW","created":"2026-01-21 15:35:24.000000000","updated":"2026-08-30 11:11:09.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fc5d1be3b94e0f97bd77543c4e336dd5acaabb59","_number":974096,"virtual_id_number":974096,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-30 11:11:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-21 18:37:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-17 15:56:56.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"81d32ff9fe4dbae011f224f324570a3f7b1edb30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-21 15:35:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"764dbda5ab9a175523b8529b10a799a490c8ea03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 18:37:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/350aed8e5e0d4843bc9e67ddd796f5c1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c00fba4f11cc4bd1b4b90bb13a94b637 : SUCCESS in 2m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e489949d31cd41a899d46cb23a9d0f4b : SUCCESS in 2m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e3918de1bad4203b16f7f091124e795 : SUCCESS in 1h 43m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaa75d845c3542a39c42a701230e666d : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ce4085609774b99ba6958d36d0f0cd7 : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8846318705a442c7abeaf3ea216458d0 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/923fae0d59d446da9424a1d6d68f7cd1 : SUCCESS in 5m 04s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/d5d6d45087f14a21bb50ae66754debd3 : SUCCESS in 9m 49s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/c965fde58f5a490c971a9e4e0c3b5136 : SUCCESS in 2h 59m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ea5e2db8c16b1f6506439e02687493292a01bae7","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-17 15:56:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"84582eae89ca59b6cae1aa70e8aaac493ce9bf35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-17 16:53:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e342e0fcc1e60a6fcbda0c4b7b37392ac14cc01c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-17 16:54:13.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8ecbfdb98f6c10e109d16dfd5a92867b8bf6be93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 18:53:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea0add4a932a4159a4cadb2e1e859110\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f9aa8c3b1b6245e5ac0ae6a6a9360aa6 : SUCCESS in 2m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/47d5a6f29ffa458fa65eddcfb8ba1d7d : SUCCESS in 1m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/556f54aea9c349d1a7d849a082017010 : SUCCESS in 1h 00m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff4e8c90e096478cb66536f03ac02593 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ff66d3f6a9b4173ac5a5559e90d6ad8 : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a4bfd9635b594ad4b49f2818040d9c26 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb18473639154f91810182cc38a0c60b : SUCCESS in 3m 26s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/adbb6922cbd74091860634ec9a063d4e : SUCCESS in 7m 35s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/b15f1f593f874c1784a1ed5ab580e60e : SUCCESS in 1h 41m 36s","accounts_in_message":[],"_revision_number":2},{"id":"32c46731792bb3678212ddfb92c3be090c8f4a1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-18 17:52:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"41be8fe30e4752e249181307c63ec412ca9128e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-18 20:32:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bfad16ce1534ef5ae4b3dda25d1c790\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/961ac06c406a4bafb0203d9d15bbc0a6 : POST_FAILURE in 2m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/687584ebf9424e99b7d959ac1c4301f5 : POST_FAILURE in 2m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cafd4cbb6382481887518556b7e0e638 : POST_FAILURE in 1h 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecc7380174004f0c886e761c8964eb3b : POST_FAILURE in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d8ea3cd6b1334f039597e0807aa6044a : POST_FAILURE in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a27fc7963ef2444fa9c9bc1d2c3c93c8 : POST_FAILURE in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/75b8a8d4af7e4442bd973fbd5d799ac3 : POST_FAILURE in 8m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71767bd633954d8e826f44f443e17637 : POST_FAILURE in 1m 44s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/83ad03c8cb59432d82f1d561336c5063 : POST_FAILURE in 6m 44s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/cfe609645dcc4c909cc0bd2222efbb0f : POST_FAILURE in 2h 31m 54s","accounts_in_message":[],"_revision_number":3},{"id":"439e4baac1b4adf4ac813ef99cd7604b61a585a6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-27 17:55:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"039b872679e8586c88be8bed40b4bb8e1269f2d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-27 19:33:41.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/32370f3839de4456b4c8e1bdff3af9d2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d38890ddca97470a85ac0d68f4a07a51 : SUCCESS in 2m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c334ec404e4742c7843c367b0a949cef : SUCCESS in 2m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ee866f1402ca4b86a97b897ecb30cf79 : SUCCESS in 57m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee701a0b51364480ade5394182b07f1d : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ae2e8d61cf24cd3a8d57f07d1a670bb : SUCCESS in 5m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/14a5cf2b7ff445f098a2003f799ef701 : SUCCESS in 3m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9c71a756846847bfa1a35e605017ffc0 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6473fbabc4144ea3aab01724349ce020 : SUCCESS in 3m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7e3b4b6c0e6494db5766b021e410a4a : SUCCESS in 3m 40s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/5fe3d31bbbb6456d8e7c2cd0caff6978 : FAILURE in 6m 39s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/7801544ddff14b9e899ddb960008f2e3 : SUCCESS in 1h 36m 58s","accounts_in_message":[],"_revision_number":3},{"id":"4b2b9adcf80b101638f0a204969bd31b11e46cb4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-24 12:31:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"976234d86c4376e5d24b0727f07ea3be35de3a18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 15:02:05.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b720ba232b34fcd969ab455b5715138\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f692af4ed555433db1b6e5a32aaf010e : SUCCESS in 2m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c33b4dc88cfe4f328515e557e7c53759 : SUCCESS in 3m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71e5aa19424949f09adca82f56983ad7 : FAILURE in 1h 45m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c44519239b4e4f96a620592b14967faf : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ccec08c516064f5398486a86fb261554 : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d07b6e3c3cfc48fc8b9f0f8119a95b1b : SUCCESS in 4m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9fae8497e94346fd9d86887e64b53c59 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13451d43f0ee41848aa4f892f6ea145e : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d457a3290a148c797ef67d1542b3672 : SUCCESS in 3m 36s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/9468a48fd9c148be9092c5601c0c95f5 : FAILURE in 7m 12s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/0594bfdca7a14dce8ceea02a720d2c34 : FAILURE in 2h 26m 29s","accounts_in_message":[],"_revision_number":3},{"id":"c77deac5215f35cd1b9f0e1cfbf52722825626fa","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-24 15:20:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"27c7b8b5c85f73cf261b3b4c928cdb47ebc21c50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 18:11:34.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d560ddb43b946988c647d1224a254d4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/88a7baad77984eafb8b1ba2c96d5d122 : SUCCESS in 3m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b521899d89534c10a7928a7891122915 : SUCCESS in 2m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/012ea78ebb3b4859826223f1caaa7b00 : SUCCESS in 1h 52m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfb2514e29b04cf899c8802e435fca8c : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc6fa218a7e542cabff48e13632578b2 : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/55749ff45b354b44875ac96b442604b5 : SUCCESS in 4m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/61f49a648ae345ceaca6f9e2ab270406 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47ec6fba7185466f8eedca204fc0e25d : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7560f5c90a548b7a931485d34fea956 : SUCCESS in 3m 56s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/71f9c1568ab341f186401105faf69487 : SUCCESS in 7m 14s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/6e86b3a2bb764589934ced8371e5f916 : FAILURE in 2h 21m 50s","accounts_in_message":[],"_revision_number":3},{"id":"4aaa8cf210a91abbc6c88a6300dc5d05b4d8364c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-24 18:16:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"667d1a1cf12f652ca7cd5cc94b70230fd5e5374f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 20:52:23.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc5845335d1b43d0a3cf69d070d124c7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aae6c9654a4d4f46b77e963a9c999aed : SUCCESS in 3m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/496c72d5e8b64cc4b70ea24e32741563 : SUCCESS in 2m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64c5cb2a2b5048d2b2624db57035a224 : SUCCESS in 1h 24m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bd3e214c28b4217b756ad664d9b4c6c : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dbae6179521849b78ec924c0329b0664 : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/663aad2ed9764ec0afc4e6ef751bd0c7 : SUCCESS in 3m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eea1618f55fb414f9145aa47e2dbcccc : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bae6830e9fab4cb381779c10c3436b07 : SUCCESS in 2m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aff1a59470564ef6851f61e7d4a7e535 : SUCCESS in 2m 23s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/f3bebff845bd415c8f8ff28863d1dd7b : SUCCESS in 5m 01s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/482269f01dcf4cc5a59c3cbe4510a0a2 : FAILURE in 2h 34m 20s","accounts_in_message":[],"_revision_number":3},{"id":"23b4ff816a8a2f5e1516377f1055b9229d2e22e9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-26 16:35:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f1cafefb78a1fdd4df86e195742a3ef142db7b00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 18:43:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5bf0469af0f5464299e7bba851829ed3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3790e26d06d6489ea4b4f136690bd74d : SUCCESS in 4m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fc7407d5f4624d75828b974283d8993d : SUCCESS in 2m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8faf7e668cb74ca2898c4a1fca392efd : SUCCESS in 1h 43m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b220b56872124441bd7f11be2007c37e : SUCCESS in 2m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47895e3f42574e2a88bc21e738d9ef74 : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0d35ab7b3ecc4e7e8a0c7dba5e4d3698 : SUCCESS in 3m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a1452f866b0f4548a448b0b87e10c442 : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e96e39dba3f44759bdd15a22f423b9ae : SUCCESS in 3m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70cae6e2d05c4ef2ba642cc6830e82ba : SUCCESS in 3m 35s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/72cac040086241f4ae988de19a2d9542 : SUCCESS in 7m 37s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/bda93a48b5e943298eed6e42cbd90f2e : FAILURE in 1h 53m 35s","accounts_in_message":[],"_revision_number":3},{"id":"0c044a5df130ca98aca04de4839904e04aae8157","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-29 16:57:13.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"9fa4eaf1437072315e58a130197ee69132aab117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 18:01:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/49645faa733f43e28fdf8e8b988d3745\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/20c07d9c4e3446be86385ec4428d4a18 : SUCCESS in 2m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/af7f240647074e0a9f731e52ed3cf428 : SUCCESS in 1m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49d05b69d41c4a09b2c35a6da30b0a1b : SUCCESS in 1h 01m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dcfa2ce876e4a61b6a0618971714798 : SUCCESS in 2m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b63bc715f7b497cbd715aedcd9fd13e : SUCCESS in 5m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5838d715f694d6fb1e6be63d40b7624 : SUCCESS in 3m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ddc4e09541d642a79e6d25e642e36aea : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b28ef0ebae374e7e95c219f77e1b4e18 : SUCCESS in 2m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0190d7c9e971424c96d15ddd02011535 : SUCCESS in 3m 16s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/6f8bd344f1a247b0958a67a815e51fba : SUCCESS in 6m 17s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/f5e6ea08e412498ca80fdf6f35302d47 : FAILURE in 7m 13s","accounts_in_message":[],"_revision_number":4},{"id":"30f73f729e183cbaa58658a5780377b95b9ae264","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-29 19:16:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bc6eca46436fb87f1d44c70d40c2e5a8591fac60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-29 19:25:54.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"06182bc65193e9102a84eb0270f74111c7da5e9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 22:03:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5bdb89e3e9844e8193e5bf1be2297bec\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3acd9a2a246340fd9a16cf7dd28ce64a : SUCCESS in 2m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7112272ef0ca4bf0ac9e7f9afd17854f : SUCCESS in 2m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/81f49b9932e54fd9b241a4f398a876df : SUCCESS in 42m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fade87c5763477c90a433aafc3e6803 : SUCCESS in 2m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d78ed954496410daac53b22d69b401e : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/055928bd68d6438e88c51c1c87cfe14d : SUCCESS in 4m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/acf8f1699722483ca2c107d67f2ff985 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79a88e73327a460bb8506f53593c956a : SUCCESS in 2m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b8250b0165b4ce09fde32f80f244453 : SUCCESS in 2m 39s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/1dbf96c8be5e4074a1d80340ca69a474 : SUCCESS in 6m 28s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/627bf1e43c0e440f97dc5a07e8ffbf48 : FAILURE in 2h 31m 44s","accounts_in_message":[],"_revision_number":5},{"id":"1aaf1312d5ed8bd11246374dd7eae5b73ebc1942","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-30 04:31:45.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"8597c475f3e53224e8f58e206e1b58f796c9e2c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-30 05:42:15.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"423a328ae1ef25f2c4f7f7b4dc07b610305c25ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 07:50:52.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0aaff82de954b9183e18b456edb6c37\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/124cf5805a70478fa0c2347c4e2bab27 : SUCCESS in 4m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e6b5edc76b4740e5b18ec99a220b71c0 : SUCCESS in 3m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ca93ea76469d4db49aaefb17b5663a52 : FAILURE in 32m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d683ff2edef44962899a96710317002b : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f9eb5553a9c349c99040e6bf50846876 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e926ccea21be40e09e4660351a38975c : SUCCESS in 4m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/167970d0a16b42248102f5ef08aaa4f6 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3cb3112339f4346b01c35dae6f1db63 : SUCCESS in 2m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88ea1f384a0d4dd9a176100f0d9b92eb : SUCCESS in 3m 45s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/f2be66e129f54b5da67aafab891971b3 : SUCCESS in 7m 30s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/0dc9bddb7e764cb9bbe691bb244e3c04 : SUCCESS in 2h 01m 47s","accounts_in_message":[],"_revision_number":7},{"id":"76ea70519774e01b23e20dc7898af3c3687f755a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-30 08:42:08.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fc5d1be3b94e0f97bd77543c4e336dd5acaabb59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 11:11:09.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72b21671b0244e9aa5d980ad837f4ea3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/51a462669bc04d1b85cace6e502ed5fd : SUCCESS in 2m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ac516631c43544dbb8c9706ddde5abeb : SUCCESS in 2m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8482fb973db649aab3609459d9224147 : SUCCESS in 1h 01m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/696e3642f50f49a2b5f2a009ff8dd8dc : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2c9bfeaef8a41288c11e0ee7dfbd153 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2240439c35fa453cbc655f8aaab82050 : SUCCESS in 3m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8c52a958c73d453ab2697d9232c2a2b0 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66755a57a54e4be78cd597c402436208 : SUCCESS in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ba4b8b1a7d3491e8e07849007a271d8 : SUCCESS in 3m 37s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/526bd624fad34a6fbb2fc74433f94f55 : SUCCESS in 6m 52s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/1bc219432c8147ae80098fac4e795e22 : SUCCESS in 2h 23m 52s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"d27127c755410ce94da416bbc5cdfb005f0adfee","revisions":{"823e98689185299906071110b9126ecf64526cac":{"kind":"REWORK","_number":1,"created":"2026-01-21 15:35:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/974096/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/96/974096/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/96/974096/1"}}},"commit":{"parents":[{"commit":"7d1cb7963098f5421cc1784ddbd7c1a371dab947","subject":"Declare Python 3.13 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/7d1cb7963098f5421cc1784ddbd7c1a371dab947"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-21 15:34:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-21 15:34:49.000000000","tz":540},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: I57e70989f6bcab586a28ec150ac6b6c5d8a86918\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/823e98689185299906071110b9126ecf64526cac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/823e98689185299906071110b9126ecf64526cac"}]},"branch":"refs/heads/master"},"3febe23e6488eba88f47936e20b2b9164de9e173":{"kind":"REWORK","_number":2,"created":"2026-02-17 16:53:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/974096/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/96/974096/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/96/974096/2"}}},"commit":{"parents":[{"commit":"7d1cb7963098f5421cc1784ddbd7c1a371dab947","subject":"Declare Python 3.13 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/7d1cb7963098f5421cc1784ddbd7c1a371dab947"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-21 15:34:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-17 16:53:16.000000000","tz":540},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: I57e70989f6bcab586a28ec150ac6b6c5d8a86918\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3febe23e6488eba88f47936e20b2b9164de9e173"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3febe23e6488eba88f47936e20b2b9164de9e173"}]},"branch":"refs/heads/master"},"c7c1b7f4853855e161028bc2ecc45338fc4a3ff2":{"kind":"REWORK","_number":3,"created":"2026-04-18 17:52:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/974096/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/96/974096/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/96/974096/3"}}},"commit":{"parents":[{"commit":"0e3cdfb2266e500d03c2a7127a8b7ae3a558cbdc","subject":"Add pyproject.toml to support pip 23.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0e3cdfb2266e500d03c2a7127a8b7ae3a558cbdc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-21 15:34:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-18 17:51:58.000000000","tz":540},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: I57e70989f6bcab586a28ec150ac6b6c5d8a86918\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/c7c1b7f4853855e161028bc2ecc45338fc4a3ff2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/c7c1b7f4853855e161028bc2ecc45338fc4a3ff2"}]},"branch":"refs/heads/master"},"a9be0e4b99e71d6255b36b8a229c4fe115c416b7":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-08-29 16:57:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/974096/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/96/974096/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/96/974096/4"}}},"commit":{"parents":[{"commit":"532a6528b25aed021cf7dac59b405a247d2a9e3e","subject":"Fix wrong section for do_secure_hash option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/532a6528b25aed021cf7dac59b405a247d2a9e3e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-21 15:34:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-29 16:57:13.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: I57e70989f6bcab586a28ec150ac6b6c5d8a86918\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a9be0e4b99e71d6255b36b8a229c4fe115c416b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a9be0e4b99e71d6255b36b8a229c4fe115c416b7"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"0e3cdfb2266e500d03c2a7127a8b7ae3a558cbdc","ours":"c7c1b7f4853855e161028bc2ecc45338fc4a3ff2","theirs":"532a6528b25aed021cf7dac59b405a247d2a9e3e","merge_strategy":"recursive","contains_conflicts":false}},"e04c3536cb4b08d1a55d1c6764f943dcb95d123f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-08-29 19:25:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/974096/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/96/974096/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/96/974096/5"}}},"commit":{"parents":[{"commit":"cefeb41e39b0bc2a47aa575ea5aa0c6ba871198a","subject":"tempest: Configure do_secure_hash option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/cefeb41e39b0bc2a47aa575ea5aa0c6ba871198a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-21 15:34:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-29 19:25:54.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: I57e70989f6bcab586a28ec150ac6b6c5d8a86918\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/e04c3536cb4b08d1a55d1c6764f943dcb95d123f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/e04c3536cb4b08d1a55d1c6764f943dcb95d123f"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"532a6528b25aed021cf7dac59b405a247d2a9e3e","ours":"a9be0e4b99e71d6255b36b8a229c4fe115c416b7","theirs":"cefeb41e39b0bc2a47aa575ea5aa0c6ba871198a","merge_strategy":"recursive","contains_conflicts":false}},"87bd8e0aff279198890085e5c52e87a44140ba74":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-08-30 04:31:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/974096/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/96/974096/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/96/974096/6"}}},"commit":{"parents":[{"commit":"e51f7d2217cc26151db62926241f8dc165d97465","subject":"tempest: Configure do_secure_hash option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/e51f7d2217cc26151db62926241f8dc165d97465"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-21 15:34:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-30 04:31:45.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: I57e70989f6bcab586a28ec150ac6b6c5d8a86918\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/87bd8e0aff279198890085e5c52e87a44140ba74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/87bd8e0aff279198890085e5c52e87a44140ba74"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"cefeb41e39b0bc2a47aa575ea5aa0c6ba871198a","ours":"e04c3536cb4b08d1a55d1c6764f943dcb95d123f","theirs":"e51f7d2217cc26151db62926241f8dc165d97465","merge_strategy":"recursive","contains_conflicts":false}},"d27127c755410ce94da416bbc5cdfb005f0adfee":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-08-30 05:42:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/974096/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/96/974096/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/96/974096/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/96/974096/7"}}},"commit":{"parents":[{"commit":"658dca5cf839539f23e8d9e56a69d6cf30eda77d","subject":"tempest: Configure do_secure_hash option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/658dca5cf839539f23e8d9e56a69d6cf30eda77d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-21 15:34:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-30 05:42:15.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: I57e70989f6bcab586a28ec150ac6b6c5d8a86918\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d27127c755410ce94da416bbc5cdfb005f0adfee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d27127c755410ce94da416bbc5cdfb005f0adfee"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"e51f7d2217cc26151db62926241f8dc165d97465","ours":"87bd8e0aff279198890085e5c52e87a44140ba74","theirs":"658dca5cf839539f23e8d9e56a69d6cf30eda77d","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
