)]}'
{"id":"openstack%2Fglance_store~978367","triplet_id":"openstack%2Fglance_store~master~I9256cbeedeb09fdac90b630aaf3b95f8d763af06","project":"openstack/glance_store","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-03-13 18:43:59.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I9256cbeedeb09fdac90b630aaf3b95f8d763af06","subject":"Remove the last eventlet import","status":"MERGED","created":"2026-03-02 02:54:52.000000000","updated":"2026-03-13 20:05:59.000000000","submitted":"2026-03-13 20:05:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978367-eventlet-removal","meta_rev_id":"0d1d57aee71b8e46ded4cdc1933d71ed3ac660e0","_number":978367,"virtual_id_number":978367,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-13 20:05:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-03-05 14:22:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-03-13 18:43:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2026-03-13 18:43:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-02 05:01:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-05 14:22:53.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"c57cb9d8179e40fe13786e91e5b7542f997e9adb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-03-02 02:54:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b7832523d0c7a96892bda69961b32a9e0b3c713","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-03-02 02:55:46.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3943b2087fa13f2b6aa16c23d677ef03262ab29d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 05:01:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e07c11f11a7b4d5fb94d96342191c046\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1460245c24d849189d945c8764ad6383 : FAILURE in 1h 04m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/526cd1fcbf884f138bf18dce8170d328 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85e1f2edc0bf480a8ed7925a07a69d39 : FAILURE in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/edb6a238ea44443082aa5888fe8c5ae6 : FAILURE in 4m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/205861951bcd4a159bacd25b9ed99191 : FAILURE in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/851b6209356c41d69f547509950dc95b : SUCCESS in 3m 19s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/51e27f4a2da14a84939c2f52d841da86 : SUCCESS in 11m 10s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/0885ea8dace6437091fe6c7f6928b561 : SUCCESS in 2h 04m 52s","accounts_in_message":[],"_revision_number":1},{"id":"2f18ba75cfe202cc1d17bb43297b715f68c7c235","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-03-02 14:35:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3c62e781136613015c48515637041e0c7c9c5ca4","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-03-02 14:36:15.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"3be2e2ad6da50b01bae9d144076b5817f3a11aba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 17:21:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cae6953e12814b199621a84e89ec31d4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5703cf61ca9d4b158a59caf29c3396b2 : SUCCESS in 1h 27m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/354efdc4506c4072921a684be1018856 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02f979e31e714e7cb97107face8f3653 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5d64d91cb16f4707a43939aebfa18331 : SUCCESS in 3m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/30d09905edb1409c9faa69da38fde451 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e60abe370dce4b36b7e34cd4aca4a70c : SUCCESS in 2m 50s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/4aa2cb5a10bd4dee943e0b36245a28db : SUCCESS in 7m 03s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/11c583f1b4cd4a54b456eb8a181eaf77 : SUCCESS in 2h 32m 35s","accounts_in_message":[],"_revision_number":2},{"id":"6babbf81253efe27d340b70ef4b63b7283c7d3c4","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-03-03 16:40:38.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"a0d8d6f908df3ccc0f05d89074c70ce7e1add7bd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-03-05 14:22:53.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b370d428c6d77dfa81a4065509596145521a8c65","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-03-13 18:43:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b5bb73ee33bfdd426a23b06cbe0ea002ebaa2f53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 18:44:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d64bbf134b1125cda3baa0f33405d2cf935b4ed2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 20:05:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2007ffddad144d88b72236b458da23d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e8e13a5983347b3b5e65620375fca1f : SUCCESS in 1h 19m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97b2b3b1542746f59b098cfaa172083d : SUCCESS in 1m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1dad370cc5e24c48910c693a9447a5f0 : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b3d5353f74e24d078cfc064be4c05092 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bab5e564c99e43fda4cb8b6667a70d1e : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":2},{"id":"f2eb50f39e7fe9bde8b808487b254ec54c9e2721","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 20:05:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0d1d57aee71b8e46ded4cdc1933d71ed3ac660e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 20:05:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/819d7e45cf244279a21643a55ee0570b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78dcb210f9b0418a9b0a35c8261e560d : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c3e4d0a8c66679a2056b440564d3236161a8d21d","revisions":{"3938b75a92d8aa7945941e050bc263674cbb0b8a":{"kind":"REWORK","_number":1,"created":"2026-03-02 02:54:52.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/67/978367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/67/978367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/67/978367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/67/978367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/67/978367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/67/978367/1"}}},"commit":{"parents":[{"commit":"032d273acc57f9010123f184cb112979f37391a5","subject":"Merge \"s3: add options to control checksum calculation/validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/032d273acc57f9010123f184cb112979f37391a5"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-03-02 02:51:32.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-03-02 02:54:44.000000000","tz":60},"subject":"Remove the last eventlet import","message":"Remove the last eventlet import\n\nChange-Id: I9256cbeedeb09fdac90b630aaf3b95f8d763af06\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3938b75a92d8aa7945941e050bc263674cbb0b8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3938b75a92d8aa7945941e050bc263674cbb0b8a"}]},"branch":"refs/heads/master"},"c3e4d0a8c66679a2056b440564d3236161a8d21d":{"kind":"REWORK","_number":2,"created":"2026-03-02 14:35:47.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/67/978367/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/67/978367/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/67/978367/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/67/978367/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/67/978367/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/67/978367/2"}}},"commit":{"parents":[{"commit":"032d273acc57f9010123f184cb112979f37391a5","subject":"Merge \"s3: add options to control checksum calculation/validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/032d273acc57f9010123f184cb112979f37391a5"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-03-02 02:51:32.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-03-02 14:35:39.000000000","tz":60},"subject":"Remove the last eventlet import","message":"Remove the last eventlet import\n\nChange-Id: I9256cbeedeb09fdac90b630aaf3b95f8d763af06\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/c3e4d0a8c66679a2056b440564d3236161a8d21d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/c3e4d0a8c66679a2056b440564d3236161a8d21d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
