)]}'
{"id":"openstack%2Fglance_store~985234","triplet_id":"openstack%2Fglance_store~master~I2a93174057c57796ee593fa440d46c1ed74a499a","project":"openstack/glance_store","branch":"master","topic":"pip-23.1-support","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-01 15:14:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2a93174057c57796ee593fa440d46c1ed74a499a","subject":"Add pyproject.toml to support pip 23.1","status":"MERGED","created":"2026-04-18 17:51:29.000000000","updated":"2026-06-01 15:15:46.000000000","submitted":"2026-06-01 15:14:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"985234-pip-23.1-support","meta_rev_id":"c97548646b3961195a288e7b5cc64043f8e698aa","_number":985234,"virtual_id_number":985234,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-01 15:14:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2026-06-01 13:55:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2026-06-01 13:55:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-18 19:20:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-01 13:55:21.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"6c43955c34dc954643feb2a17ecbb977ac302c2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-18 17:51:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4412c4faa9fe4d3f08161eaf93d66fbc31f94e6e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-18 17:52:41.000000000","message":"Topic set to pip-23.1-support","accounts_in_message":[],"_revision_number":1},{"id":"cc7c58eaae939e2960d4e659fad32f5e4a537de6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-18 19:20:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/df64495ef319442ba8295a43438da76f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f70d2352457d40e9924d68e462db7205 : POST_FAILURE in 1m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b36763fb5045473a8029868d47f0ad45 : POST_FAILURE in 1m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f83c6ab42fd4c848408c8ce524a2c4c : POST_FAILURE in 40m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95ac78d5adcc49baa1efac6645c60f7b : POST_FAILURE in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c793dc46fc494b7187c462cbb8bac607 : POST_FAILURE in 2m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1176ef05900846378fc356fd6dccc458 : POST_FAILURE in 2m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9a21ae9acc524f4d8597bd6ac993df8c : POST_FAILURE in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08a55892c32b4833951bfd9e1fb3dd5c : POST_FAILURE in 2m 27s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/cfa199ad4f5048a6b05dd350273ec851 : POST_FAILURE in 8m 16s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/dca1c775b5934557913b2d8dcc91b79e : POST_FAILURE in 1h 27m 47s","accounts_in_message":[],"_revision_number":1},{"id":"fd5c27c7ff5c58f60cd235c54d48f5d16221e3cf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-19 05:23:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1fb4770f6861b5cf07183cfdf556c980a249798b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-19 08:33:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/54d65c32167947a89b08da1735472278\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e0ed628641ec467cbcd75fabe2a7ed81 : POST_FAILURE in 1m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/062b273d12b64f7581db4a757626e8c2 : POST_FAILURE in 1m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0c49779e4d9e4c958d1cf3686b1c134d : POST_FAILURE in 57m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c87e252397b4f1bb54889eb38e0c57d : POST_FAILURE in 2m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34caf54263e747bb9196d683b9053133 : POST_FAILURE in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/05580c0befb0434fa2d721e34b216cb5 : POST_FAILURE in 4m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e9978592c8c346fea9f95f9cd0873a07 : POST_FAILURE in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa54b134e8e348c59fb572025f15f096 : POST_FAILURE in 3m 59s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/7f3f81c2606b4cabab8d069a31462150 : POST_FAILURE in 8m 27s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/a2011c2e97944f15a2d73661152c7f2f : TIMED_OUT in 3h 03m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7a1c0b4b916c10ba01a941ad994d5b91b2f8bd6a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-20 13:24:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e309fe2f8ec06823a366f45bc4bfc7ce943d0fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 17:14:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bfd1354ac0a44c7bf8e61c68cabe891\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d30f9693046947d0ae8522b1bfc1c94a : SUCCESS in 3m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a05f41aef23b42079dec78910e7d659c : SUCCESS in 3m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88514e0897b64203ae108e3b74f37305 : SUCCESS in 1h 52m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6076a37f5f494a61a9d3f6ba519247bc : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b294f5775f644c989e33dc1a7f99779 : SUCCESS in 5m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10850bef9270485f9ded4b80fece89de : SUCCESS in 5m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2d1f844ffce446f38bfe717c617ef379 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dacfc88a3184a8f9f379b39125c40df : SUCCESS in 3m 44s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/ea9d08d474244db2b9f48a9fb29c5cf6 : SUCCESS in 7m 37s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/8373fa573670478f90a71b6089f75837 : TIMED_OUT in 3h 05m 47s","accounts_in_message":[],"_revision_number":1},{"id":"51599fa16489e2f239a5f143e660b01d8e95cbb9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-21 03:52:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"438c4caeb621805c827db4ae95c37270a51c1119","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 05:30:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/907b8820db3c42ba97009679cb6194ad\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a122db4c687e4a9eaa91449dbf5068fd : SUCCESS in 1m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b1407cf9adb044c0b07802c8ca8f9029 : SUCCESS in 3m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c93d55ea493a4320b8745f146d153d88 : SUCCESS in 1h 21m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dee5495a779946b2a8ce3bf315f1d33b : SUCCESS in 1m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/416e5284cf264dcbba5e86d430ad644c : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fc820106c1d24046962ff733051684c6 : SUCCESS in 2m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/36ab794861ee477fbf30f72cafe66a5b : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bc76984fd0c4569ae8dac92d4a960c1 : SUCCESS in 2m 06s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/a930b639493e4c6bb3194dba3a0102a0 : SUCCESS in 4m 45s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/ae080df8a143484cb4db8556658fa143 : SUCCESS in 1h 36m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5d7881c900883e288fee2126ac895bd2d3ad3d1d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-06-01 13:55:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a7f777d915e0da93692c5f86ba5dcb38f2b33b13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 13:55:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6206b2007661b5f319185cb778caca1e4ee6075d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 15:14:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/532a2253fdfd412eaf8b5491973bbf12\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1389e06e42804d949902d1e0757cfbbd : SUCCESS in 3m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6a7285d34f574e338caac77fe0832589 : SUCCESS in 1m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d501c3688674cd19329b9545c6824a7 : SUCCESS in 1h 17m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4701afb76f940d9be4ec9d4e6b19f8b : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3a42b2ac97e4faf8d7228135b8ec565 : SUCCESS in 6m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/452d2cf317724f5fa0d67359fcc0b876 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1d5b77cb4e342b2aeb5c837e0eb455a : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5206dcb7914d48a6be1c73237b056bfb : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d57a15792414ecb4bf8dd6930d3580fdd82534c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 15:14:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c97548646b3961195a288e7b5cc64043f8e698aa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 15:15:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4103c58f709e48c9a238263164e39dbf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9041134ad14d43199f7ad986e8fc5097 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e3cdfb2266e500d03c2a7127a8b7ae3a558cbdc","revisions":{"0e3cdfb2266e500d03c2a7127a8b7ae3a558cbdc":{"kind":"REWORK","_number":1,"created":"2026-04-18 17:51:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/985234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/34/985234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/34/985234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/34/985234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/34/985234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/34/985234/1"}}},"commit":{"parents":[{"commit":"f32b259befa85ad8a2ee5948341c8389e445ddd3","subject":"Fix Swift multi-backend config for downloads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/f32b259befa85ad8a2ee5948341c8389e445ddd3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-18 17:50:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-18 17:50:48.000000000","tz":540},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects\nthat do not have pyproject.toml and now uses a pyproject.toml\nwhich is vendored in pip.\nTo address that, this change adds the minimal pyproject.toml\nto enable pbr to be properly used to build editable wheels.\n\nThis is required to support installing devstack on\ncentos stream 9 and related distros with GLOBAL_VENV\u003dTrue\nWithout this change the wsgi scripts are not generated in\neditable mode. i.e. pip install -e /opt/stack/keystone\n\nSee https://pip.pypa.io/en/stable/news/#v23-1\nand https://github.com/pypa/pip/issues/8368 for more\ndetails on the removal of the fallback support.\n\nsetuptools v64.0.0 is used to support editable installs\nvia its PEP-660 implmentation\nhttps://github.com/pypa/setuptools/pull/3488\n\nCo-Authored-By: Sean Mooney \u003cwork@seanmooney.info\u003e\nChange-Id: I2a93174057c57796ee593fa440d46c1ed74a499a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0e3cdfb2266e500d03c2a7127a8b7ae3a558cbdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0e3cdfb2266e500d03c2a7127a8b7ae3a558cbdc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
