)]}'
{".zuul.yaml":[{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"bc584b94a1853f8ae089c8e1ea8512229128800d","unresolved":false,"context_lines":[{"line_number":169,"context_line":"      - ^tools/.*$"},{"line_number":170,"context_line":"      - ^tox.ini/.*$"},{"line_number":171,"context_line":""},{"line_number":172,"context_line":"- project:"},{"line_number":173,"context_line":"    templates:"},{"line_number":174,"context_line":"      - check-requirements"},{"line_number":175,"context_line":"      - lib-forward-testing-python3"},{"line_number":176,"context_line":"      - openstack-python3-jobs"},{"line_number":177,"context_line":"      - publish-openstack-docs-pti"},{"line_number":178,"context_line":"      - release-notes-jobs-python3"},{"line_number":179,"context_line":"    check:"},{"line_number":180,"context_line":"      jobs:"},{"line_number":181,"context_line":"        - cross-glance-tox-functional"},{"line_number":182,"context_line":"        - glance_store-src-ceph-tempest:"},{"line_number":183,"context_line":""},{"line_number":184,"context_line":"    experimental:"},{"line_number":185,"context_line":"      jobs:"},{"line_number":186,"context_line":"        - glance_store-dsvm-functional-filesystem"},{"line_number":187,"context_line":"        - glance_store-dsvm-functional-swift"},{"line_number":188,"context_line":"    periodic:"},{"line_number":189,"context_line":"      jobs:"},{"line_number":190,"context_line":"        # NOTE(rosmaita): we only want the \"tips\" jobs to be run against"},{"line_number":191,"context_line":"        # master, hence the \u0027branches\u0027 qualifiers below.  Without them, when"},{"line_number":192,"context_line":"        # a stable branch is cut, the tests would be run against the stable"},{"line_number":193,"context_line":"        # branch as well, which is pointless because these libraries are"},{"line_number":194,"context_line":"        # frozen (more or less) in the stable branches."},{"line_number":195,"context_line":"        #"},{"line_number":196,"context_line":"        # The \"tips\" jobs can be removed from the stable branch .zuul.yaml"},{"line_number":197,"context_line":"        # files if someone is so inclined, but that would require manual"},{"line_number":198,"context_line":"        # maintenance, so we do not do it by default.  Another option is"},{"line_number":199,"context_line":"        # to define these jobs in the openstack/project-config repo."},{"line_number":200,"context_line":"        # That would make us less agile in adjusting these tests, so we"},{"line_number":201,"context_line":"        # aren\u0027t doing that either."},{"line_number":202,"context_line":"        - glance_store-tox-py3-cinder-tips:"},{"line_number":203,"context_line":"            branches: master"},{"line_number":204,"context_line":"        - glance_store-tox-py3-keystone-tips:"},{"line_number":205,"context_line":"            branches: master"},{"line_number":206,"context_line":"        - glance_store-tox-py3-oslo-tips:"},{"line_number":207,"context_line":"            branches: master"},{"line_number":208,"context_line":"        - glance_store-tox-py3-swift-tips:"},{"line_number":209,"context_line":"            branches: master"}],"source_content_type":"text/x-yaml","patch_set":1,"id":"326f75da_e1b96810","line":209,"range":{"start_line":172,"start_character":2,"end_line":209,"end_character":0},"updated":"2026-06-27 18:00:54.000000000","message":"expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][1][\u0027glance_store-src-ceph-tempest\u0027]","commit_id":"2743d534c8f896bc18754fc3070a6c9dddec1f95"},{"author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"change_message_id":"b49241c3d6a70a823d3c6f95fec3f53b69058b99","unresolved":true,"context_lines":[{"line_number":131,"context_line":"              do_secure_hash: False"},{"line_number":132,"context_line":"    irrelevant-files:"},{"line_number":133,"context_line":"      - ^\\.gitignore$"},{"line_number":134,"context_line":"      - ^\\.gitreview$"},{"line_number":135,"context_line":"      - ^doc/.*$"},{"line_number":136,"context_line":"      - ^releasenotes/.*$"},{"line_number":137,"context_line":"      - ^.*\\.rst$"}],"source_content_type":"text/x-yaml","patch_set":2,"id":"26da0ad7_35abd043","line":134,"range":{"start_line":134,"start_character":20,"end_line":134,"end_character":21},"updated":"2026-08-24 11:35:22.000000000","message":"Why don\u0027t we have LICENSE, here, for instance?","commit_id":"18420865d21d2d7d5f3f23181166223a2642e022"},{"author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"change_message_id":"e52ee68906d4df29229b8288bdab1f31f1a25cca","unresolved":false,"context_lines":[{"line_number":131,"context_line":"              do_secure_hash: False"},{"line_number":132,"context_line":"    irrelevant-files:"},{"line_number":133,"context_line":"      - ^\\.gitignore$"},{"line_number":134,"context_line":"      - ^\\.gitreview$"},{"line_number":135,"context_line":"      - ^doc/.*$"},{"line_number":136,"context_line":"      - ^releasenotes/.*$"},{"line_number":137,"context_line":"      - ^.*\\.rst$"}],"source_content_type":"text/x-yaml","patch_set":2,"id":"992ac8fc_f588fef5","line":134,"range":{"start_line":134,"start_character":20,"end_line":134,"end_character":21},"in_reply_to":"26da0ad7_35abd043","updated":"2026-08-24 12:33:24.000000000","message":"I\u0027ve compared the lists and update these for more consistency.\nNote that the following changes are made in this change.\n - DO not skip tests for setup.cfg, because it affects packaging\n - Do not skip tests for requirements.txt, because it affects actual runtime dependency","commit_id":"18420865d21d2d7d5f3f23181166223a2642e022"}],"/PATCHSET_LEVEL":[{"author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"change_message_id":"b49241c3d6a70a823d3c6f95fec3f53b69058b99","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":2,"id":"46ec9cf5_912cf622","updated":"2026-08-24 11:35:22.000000000","message":"I like the idea, but it\u0027s weird that we have different lists of files for different jobs when they could probably share a list. See my inline comment.","commit_id":"18420865d21d2d7d5f3f23181166223a2642e022"},{"author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"change_message_id":"abc55411cea27c296763e34b30925ade5cc77739","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":3,"id":"f6e5ed86_d503fd50","updated":"2026-08-24 15:20:12.000000000","message":"recheck https://review.opendev.org/c/openstack/glance/+/1002028 was merged","commit_id":"8ff4b18d85b990d1a0a7d66bdf62421dd00a6118"},{"author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"change_message_id":"50c59b7f85175eff0d26e303c295cb11f35ae1d5","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":3,"id":"286a2d75_6cc545aa","updated":"2026-08-26 16:35:31.000000000","message":"recheck let me try once more. The job passed in a different change.","commit_id":"8ff4b18d85b990d1a0a7d66bdf62421dd00a6118"}]}
