)]}'
{"id":"openstack%2Fglance~1000061","triplet_id":"openstack%2Fglance~stable%2F2025.1~I6fbb8a91a49e3df6fa9bb185f7fcd8b1816ad74e","project":"openstack/glance","branch":"stable/2025.1","topic":"bug/2152110","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-09-28 16:46:28.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-09-28 14:03:47.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-09-28 16:46:28.000000000","reason":"Change was submitted"},"34510":{"account":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"last_update":"2026-09-09 07:07:58.000000000","reason":"\u003cGERRIT_ACCOUNT_34510\u003e replied on the change","reason_account":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}}},"hashtags":["OSSN-0105"],"change_id":"I6fbb8a91a49e3df6fa9bb185f7fcd8b1816ad74e","subject":"Apply import URI filtering to legacy import tasks","status":"MERGED","created":"2026-08-07 00:02:16.000000000","updated":"2026-09-28 17:05:16.000000000","submitted":"2026-09-28 16:46:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"1000061-bug/2152110","meta_rev_id":"61208a7bf8e9157b2b2f951c9b1ad9ca1e1fcf21","_number":1000061,"virtual_id_number":1000061,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-28 16:46:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-09-28 14:03:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-09-28 14:03:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-07 00:02:16.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2026-08-07 01:45:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-07 02:47:32.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2026-08-13 06:36:40.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-08-18 10:04:02.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"state":"REVIEWER"},{"updated":"2026-09-28 14:03:47.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"3894ebaaa38d428f9a0c451bb6f338e0ef5d9157","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-08-07 00:02:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7c072924f2013c9402ff91ba4b0f6db953ff51d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 01:45:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/baaadba976904b08b46c4ea1f0fdad41\n\n- grenade https://zuul.opendev.org/t/openstack/build/c1bdb024264e441c90130e9e127f1ca5 : FAILURE in 25m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/109c94c8f13f4f978dbd2e9382110bb4 : FAILURE in 24m 42s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/96f2a55164ee4b55aefa914efdfe78c5 : SUCCESS in 1h 33m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e238f2e3496423884d150d91891b91d : SUCCESS in 37m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9e12b25db1246e3ba36c8e9c9316e20 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21fac2e33108446cb39dc63c466f5398 : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c335a4503a644d99a54cec9ea405f57e : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c7f6c4ca2814ab99e436864712e95d8 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d905d4d1999a43fdb397bc3e4d245779 : SUCCESS in 4m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ad21f045cd134a068f7b22cebe62af37 : SUCCESS in 7m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fc6d295c5403474eb2cf4f0a7e2a436f : SUCCESS in 5m 46s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ef6b6dc776e346fdafb54b1b580130bb : SUCCESS in 7m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/35ed40745a2d433ea6e77e19e8845fac : FAILURE in 1h 27m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a7deb245bbd64755870bbb54ffc1da28 : FAILURE in 23m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/45cd2fdbb571488b8c2404ab807a3b9d : SUCCESS in 1h 35m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a6eb03921f624a5eb29793b7a2cc3248 : SUCCESS in 1h 02m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6b63e783451446ffaadfabeaaad19db0 : SUCCESS in 1h 40m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/117d9b8ec4bf40f9b94ac82a1a495784 : SUCCESS in 28m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/77e6cf22264548f4981d195214b6614d : SUCCESS in 1h 10m 29s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/c48b10a880754170b0539674995fd062 : SUCCESS in 1h 39m 29s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b220053124bf4559830fc98e14800f48 : FAILURE in 21m 49s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/b418270a3c234f7c9c18e37e262fab28 : FAILURE in 14m 29s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f6da1126b6584be79a7620e61de19dae : SUCCESS in 19m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0084963614958812061903372996ebe30702f51e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-07 02:47:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bbdfd4090364120f9ec41e2da9d3c354694a213d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-07 02:48:29.000000000","message":"Topic set to bug/2152110","accounts_in_message":[],"_revision_number":1},{"id":"eada03e230af5772a28a5d482b54b56779bdc141","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-07 05:28:23.000000000","message":"Hashtag added: OSSN-0105","accounts_in_message":[],"_revision_number":1},{"id":"d50dbdcc59f62e438a1c8feaf1c4c5f85c83e244","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-08-07 08:53:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e53cc703709d582693a22dcbf45d69b9ba31742","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-13 06:40:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a1b15d82b8a99a906fb5fc1e9ff28b4606eb299e","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-09-09 07:07:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e8f3d0a7eae38d0449066bf66a7cb62fba1bd49","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-10 06:49:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93b46d7e6b02060e7846d071938029cf2768f949","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-09-10 07:03:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"081f477d0380d206af9a6242bf993db0e5332e0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-09-25 15:32:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f48314030c4e5f466901a2ab0d1fcb21aefd01f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 17:45:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d05ce034abb845cc875c6c749625bdea\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7e2cc6d460594d2c9d41df8422b6a72d : FAILURE in 45m 58s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f64d2e251f5649cb973022b4d27af37b : FAILURE in 1h 09m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61027944266c465ab1100b541328e42f : SUCCESS in 1h 05m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf4604b4f2f44a5592c2b6d3013340db : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df28bb7ff3834d18a2c260ab28ae5d8a : FAILURE in 5m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2ec416cfffc54ca596047acb96843cb0 : FAILURE in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c89fddb78eee4b2ead80858da63be4ea : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9852fc3c7e0c453a892e92f003131777 : SUCCESS in 9m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6484a198128f41c2a286eb8759b6678a : FAILURE in 9m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e9c0707ec74247dd955b9e8fd6fe150c : FAILURE in 12m 06s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c17a483a794441dd91244869ea916abc : FAILURE in 9m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/86eb15bb2125466799dd9f57612ca207 : FAILURE in 1h 23m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b8454367489842f7ad3660c0705ee304 : FAILURE in 1h 48m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/07c1838bca28494d9238c23b0c7cc897 : FAILURE in 2h 00m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3959b99905f44d44bea5f529aa23c0b6 : FAILURE in 2h 05m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5ce426cedbea471ca133571c21c9fc9c : FAILURE in 1h 55m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1c6a3b05e564a08aa2736505c488bcb : SUCCESS in 1h 04m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/31ef443bf1014ff79f3e7901953818e1 : FAILURE in 1h 51m 21s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/eaedc5ee2da1452ebbed83c8008938ab : FAILURE in 1h 50m 48s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8c164d5834a34afc8062d7932ad73789 : FAILURE in 36m 35s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/dadc8a5729f746b2aa3d0da3e64b05bc : FAILURE in 30m 21s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/02f95f08848e48ebac55e46e98bb1955 : SUCCESS in 18m 23s","accounts_in_message":[],"_revision_number":2},{"id":"a3b6be20f5bba1019a42faff4eb6dfa9dcaf1549","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-09-28 10:03:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"509eafbee1f528ae8f88d5bca7d5244961d651c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 13:03:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad48344d27414adcb34e8ea47d0af536\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/81e1e0f28ee140358eea919d0b5e0e31 : FAILURE in 50m 05s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/11ee77b6fda44870bdeb468a7314bcb8 : SUCCESS in 1h 34m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db8a553bdb414296a0cebd8566faf3ae : SUCCESS in 45m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba36373677f143e09286783bbf7f0cf8 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/466ec47f634e4e43bc69f7c6a5d39c9d : SUCCESS in 6m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e5394c18bab945d5ab5eb15a6fb86c9f : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da9e5eab8de44fa0ad0954a7887abf6d : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1a49bdfdf0d402ba87bfc4c70646b23 : SUCCESS in 9m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/383f4135f6af43a79e89bc0ad52c219f : SUCCESS in 9m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6ebad4b3fea149e99246075b4b4cb8ab : SUCCESS in 10m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d09eb4dfeec94ba2b945bd9c39c98898 : SUCCESS in 12m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4bc5597d306341d5b2b271a7dc0d1588 : SUCCESS in 1h 58m 32s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8a203cf693594b669f746661d120fc8b : SUCCESS in 1h 48m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c71f656c702c4332aa5e3d0bfa9d4c16 : SUCCESS in 2h 01m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fb01e85bb2cb4ac89438d8577d38794a : SUCCESS in 1h 58m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/25357ab43e99463dab0492ac1fef6b8a : SUCCESS in 2h 28m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb4e13c12442497585cf44b06b24dd46 : SUCCESS in 32m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6d3f34dcb6ac4377820645c1cc3760f8 : SUCCESS in 1h 18m 41s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/a9f1d76ae8244bdca3cc0166396a61f8 : SUCCESS in 2h 03m 49s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/6083f4b358b44cc692b282d55af3dcd1 : FAILURE in 35m 07s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/b63c6d8a485a4d81b0aea9c0eb9e2926 : FAILURE in 28m 22s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5b60bd2cf4f446b69d3f6de15bfe9be3 : SUCCESS in 17m 44s","accounts_in_message":[],"_revision_number":3},{"id":"90cc95092af612ce3f16ec1c7fe5cee74ab4e969","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-28 14:03:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9d61cff1969fcbb44aaf072cf9c4caee218cbf58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 14:04:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2c03c8246df2e2deb0f9af3d8e1f5c9eb3bd07de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 16:46:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e2b89d35c794f5db092b16d5a0a7709\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1a58b5697cce4084bd218d9a3f79843e : FAILURE in 58m 41s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c43de20bdf8f4fa7ba2bc36bd8eb0bef : SUCCESS in 1h 19m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f63dc9edde347f2a7bad1ea18698482 : SUCCESS in 1h 07m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9808cc0d123f4b6aab04b3a94afccdb6 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ce1adb8bb41490898af7af36108d6b3 : SUCCESS in 6m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc82fce3de0e45f28cfcabac3ffd0c46 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b5e06f6a60a4ffbb8ea7262a5c85172 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3716375f2dc14feb8d8614470ad66399 : SUCCESS in 9m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bfbfd369e3e14b94b0964ebd3f09916a : SUCCESS in 9m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0b3ebcd711e54291be4ff1899bc16e7a : SUCCESS in 10m 20s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c9f7733dbf9a409c8ce636c4b305423a : SUCCESS in 14m 59s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6d545d9e92b345b78c1d35cbc5d89f80 : SUCCESS in 2h 27m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/962df3e2c03d414e9552b2a95e214470 : SUCCESS in 2h 03m 24s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0390f29475d04616b7299792ba385ae6 : SUCCESS in 1h 52m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/231dd995003c454fa3d930fe0e7c8de4 : SUCCESS in 1h 01m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fec2cd535b3f429eafeaa8f10b8d065e : SUCCESS in 1h 55m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/90d14a25c51d4d94ace3485c3a063f3e : SUCCESS in 20m 35s","accounts_in_message":[],"_revision_number":3},{"id":"f254c8861498ccc43d4f060376af5084a1282e20","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 16:46:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"61208a7bf8e9157b2b2f951c9b1ad9ca1e1fcf21","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 17:05:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d78ee78b63744667bcd94ee02a95efc1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/399783512f1d4675bbea8794daf83c01 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/423601b073ce48bc894ae89bfe5d7ab9 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6d72304ba4988aaa589e55b4a15381397dfb2afb","revisions":{"3c1e6065f9d30d77d05976350910ff55686be20b":{"kind":"REWORK","_number":1,"created":"2026-08-07 00:02:16.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/61/1000061/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/61/1000061/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/1000061/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/1000061/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/1000061/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/61/1000061/1"}}},"commit":{"parents":[{"commit":"f1366dff052625bd70f7eff50d97417e114ea7da","subject":"Merge \"Fix glance service policy rule\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f1366dff052625bd70f7eff50d97417e114ea7da"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-05-12 06:53:53.000000000","tz":0},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-08-07 00:02:02.000000000","tz":120},"subject":"Apply import URI filtering to legacy import tasks","message":"Apply import URI filtering to legacy import tasks\n\nLegacy type\u003dimport tasks only ran validate_location_uri, so\nimport_from could skip the same host/port/path checks used for\nmodern image import. Route import_from through a small helper that\nreuses validate_location_uri then validate_import_uri, and\nreject bad URIs early when creating import tasks.\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/996220\nCloses-Bug: #2152110\nAssisted-By: Cursor (claude-4.5-sonnet) for tests\nChange-Id: I6fbb8a91a49e3df6fa9bb185f7fcd8b1816ad74e\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n(cherry picked from commit 42056e1b10abf01f458595487d66f83fbfa99db6)\n(cherry picked from commit 8d5c791e58ca51ed345c689566d43b563aff89a3)\n(cherry picked from commit ea7f6c3e164e93d70b5b04e1b14158975bf87042)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c1e6065f9d30d77d05976350910ff55686be20b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c1e6065f9d30d77d05976350910ff55686be20b"}]},"branch":"refs/heads/stable/2025.1"},"3b874825ebef7838c4e4c1d0a5f1a6151c5f150d":{"kind":"REWORK","_number":2,"created":"2026-09-25 15:32:26.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/61/1000061/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/61/1000061/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/1000061/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/1000061/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/1000061/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/61/1000061/2"}}},"commit":{"parents":[{"commit":"e0341c8222da563d57b495d04dda1141a5b02c1f","subject":"Try all allowed DNS addresses when importing image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e0341c8222da563d57b495d04dda1141a5b02c1f"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-05-12 06:53:53.000000000","tz":0},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-09-25 15:32:07.000000000","tz":120},"subject":"Apply import URI filtering to legacy import tasks","message":"Apply import URI filtering to legacy import tasks\n\nLegacy type\u003dimport tasks only ran validate_location_uri, so\nimport_from could skip the same host/port/path checks used for\nmodern image import. Route import_from through a small helper that\nreuses validate_location_uri then validate_import_uri, and\nreject bad URIs early when creating import tasks.\n\nConflicts:\n\tglance/tests/unit/async_/flows/plugins/test_image_conversion.py\n\tglance/tests/unit/async_/flows/plugins/test_inject_image_metadata.py\n\tglance/tests/unit/async_/flows/test_convert.py\n\tglance/tests/unit/async_/flows/test_import.py\n\tglance/tests/unit/async_/flows/test_introspect.py\n\tglance/tests/unit/async_/test_async.py\n\tglance/tests/unit/async_/test_taskflow_executor.py\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/996220\nCloses-Bug: #2152110\nAssisted-By: Cursor (claude-4.5-sonnet) for tests\nChange-Id: I6fbb8a91a49e3df6fa9bb185f7fcd8b1816ad74e\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n(cherry picked from commit 42056e1b10abf01f458595487d66f83fbfa99db6)\n(cherry picked from commit 8d5c791e58ca51ed345c689566d43b563aff89a3)\n(cherry picked from commit ea7f6c3e164e93d70b5b04e1b14158975bf87042)\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b874825ebef7838c4e4c1d0a5f1a6151c5f150d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b874825ebef7838c4e4c1d0a5f1a6151c5f150d"}]},"branch":"refs/heads/stable/2025.1"},"6d72304ba4988aaa589e55b4a15381397dfb2afb":{"kind":"REWORK","_number":3,"created":"2026-09-28 10:03:41.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/61/1000061/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/61/1000061/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/1000061/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/1000061/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/1000061/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/61/1000061/3"}}},"commit":{"parents":[{"commit":"e0341c8222da563d57b495d04dda1141a5b02c1f","subject":"Try all allowed DNS addresses when importing image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e0341c8222da563d57b495d04dda1141a5b02c1f"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-05-12 06:53:53.000000000","tz":0},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-09-28 10:03:24.000000000","tz":120},"subject":"Apply import URI filtering to legacy import tasks","message":"Apply import URI filtering to legacy import tasks\n\nLegacy type\u003dimport tasks only ran validate_location_uri, so\nimport_from could skip the same host/port/path checks used for\nmodern image import. Route import_from through a small helper that\nreuses validate_location_uri then validate_import_uri, and\nreject bad URIs early when creating import tasks.\n\nConflicts:\n\tglance/tests/unit/async_/flows/plugins/test_image_conversion.py\n\tglance/tests/unit/async_/flows/plugins/test_inject_image_metadata.py\n\tglance/tests/unit/async_/flows/test_convert.py\n\tglance/tests/unit/async_/flows/test_import.py\n\tglance/tests/unit/async_/flows/test_introspect.py\n\tglance/tests/unit/async_/test_async.py\n\tglance/tests/unit/async_/test_taskflow_executor.py\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/996220\nCloses-Bug: #2152110\nAssisted-By: Cursor (claude-4.5-sonnet) for tests\nChange-Id: I6fbb8a91a49e3df6fa9bb185f7fcd8b1816ad74e\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n(cherry picked from commit 42056e1b10abf01f458595487d66f83fbfa99db6)\n(cherry picked from commit 8d5c791e58ca51ed345c689566d43b563aff89a3)\n(cherry picked from commit ea7f6c3e164e93d70b5b04e1b14158975bf87042)\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6d72304ba4988aaa589e55b4a15381397dfb2afb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6d72304ba4988aaa589e55b4a15381397dfb2afb"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
