)]}'
{"id":"openstack%2Fglance~1001171","triplet_id":"openstack%2Fglance~master~I7afd99d559870a6770e93ea2cca73800a02aa996","project":"openstack/glance","branch":"master","topic":"remove-manage-fixme","hashtags":[],"change_id":"I7afd99d559870a6770e93ea2cca73800a02aa996","subject":"Remove stale glance-admin FIXME in glance-manage","status":"NEW","created":"2026-08-17 14:43:53.000000000","updated":"2026-08-17 17:17:52.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a08dc361c4e001eaac0a939b6f47bf89ac0d72d0","_number":1001171,"virtual_id_number":1001171,"owner":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-17 17:17:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-17 15:01:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-17 17:17:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"88034db3f785d1bd4369af078bd22667fadbcdd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-08-17 14:43:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"245a87c67f82ceccb86d9813e81e4ded9b5729aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 15:01:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3269667cd8a409d85f42d24e68c947f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3d0b58b81f6846d292a2a872c7d48e5e : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e6665beec0a44c9da3429a7f43a8d444 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c2bde97f2e2e4eb1a2f4cab102d7b076 : SUCCESS in 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a08dc361c4e001eaac0a939b6f47bf89ac0d72d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 17:17:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30ec25cb69d84ac5ab3a94aba4132e07\n\n- grenade https://zuul.opendev.org/t/openstack/build/cdc009dfd83d4d278ed3c1fc3169a2e9 : SUCCESS in 1h 04m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ebea0c96995c438d826093345638e192 : SUCCESS in 42m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/266986b949684dadaece530a8844e17c : SUCCESS in 1h 11m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27cab21853704b8c945757c7c487eb5a : SUCCESS in 37m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/067ccbe2405047aaa647d947a4108c94 : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6737bb953e0e4f21ab7699246809987b : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/529a8f94ff6a4e19a7d2d5d2cd5a8926 : SUCCESS in 4m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/65d84b37d3c74067ba5872b391f9732e : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b770aea2e16e473e9c3251186ef3b32b : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee2fa0670d9d4d76aedf0b63ba67a3eb : SUCCESS in 8m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fdd84a40a4324bb7827d36343b6d380a : SUCCESS in 5m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/89c6dc31b2de45eeb577061227f20b68 : SUCCESS in 6m 33s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5c6c693047da4803bfcf070d5b34a8ab : SUCCESS in 1h 41m 01s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/08162790682445e6b346df7f3a8d4603 : SUCCESS in 1h 45m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a1041953970248d398d2a21f1785501a : SUCCESS in 2h 16m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b5e6dbf054b4439a4d2cb9d5e85cfb4 : SUCCESS in 25m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/889fdd8b7f2d4110933a4247234b626a : SUCCESS in 2h 01m 12s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/ed4102914b634ce1a64a1588dbbb2853 : SUCCESS in 1h 43m 39s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/08af1433656d424e845102f57514f1ad : SUCCESS in 35m 44s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/d5031d4912df41bb8a534b87a8943b35 : SUCCESS in 48m 10s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"506f42803b383dc4ea7f63f699833eb0961681e1","revisions":{"506f42803b383dc4ea7f63f699833eb0961681e1":{"kind":"REWORK","_number":1,"created":"2026-08-17 14:43:53.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/71/1001171/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/71/1001171/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/1001171/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/1001171/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/1001171/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/71/1001171/1"}}},"commit":{"parents":[{"commit":"4f7a19df0cdad9cb50e92bc8a69be13543c4a3b5","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4f7a19df0cdad9cb50e92bc8a69be13543c4a3b5"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-08-17 14:38:30.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-08-17 14:43:27.000000000","tz":180},"subject":"Remove stale glance-admin FIXME in glance-manage","message":"Remove stale glance-admin FIXME in glance-manage\n\nThe FIXME in glance/cmd/manage.py was conditioned on a future \"glance-admin\"\ncommand that was never created. The project instead standardized on\nglance-manage as its single management CLI.Remove the stale comment\n\nChange-Id: I7afd99d559870a6770e93ea2cca73800a02aa996\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/506f42803b383dc4ea7f63f699833eb0961681e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/506f42803b383dc4ea7f63f699833eb0961681e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
