)]}'
{"id":"openstack%2Fglance~1001296","triplet_id":"openstack%2Fglance~master~I18b4b51a179b86c8e12c142cada16cc63bbc3539","project":"openstack/glance","branch":"master","topic":"chore/update-unit-stage-content-type-test-comment","attention_set":{},"removed_from_attention_set":{"34510":{"account":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"last_update":"2026-09-10 14:12:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I18b4b51a179b86c8e12c142cada16cc63bbc3539","subject":"Document config dependency in stage test","status":"MERGED","created":"2026-08-18 09:59:58.000000000","updated":"2026-09-10 14:14:44.000000000","submitted":"2026-09-10 14:12:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1001296-chore/update-unit-stage-content-type-test-comment","meta_rev_id":"ad9dd3a37d902290761b2ea01ca8f5edfc386303","_number":1001296,"virtual_id_number":1001296,"owner":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-10 14:12:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-09-03 14:21:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-09-10 12:58:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2026-09-10 12:58:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-18 10:13:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-18 11:14:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-03 14:21:43.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2026-09-10 12:58:27.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"5567bd706bd106443f776b3bf43673ea7f13cf9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-08-18 09:59:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1da82d3af806b93d0dc8410d493a32ed4e90f78d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 10:13:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2de912c4322644008225afd2e7a1f3bd\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/26e0e9a1b19546e5bf4cf2f71798e372 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d614db6a906c4bbdb5fb3c63da1ad205 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d0d8618626174ff887efa3e3c6ba87ca : SUCCESS in 8m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b2793f3a507dfb91f6647f34b376a01eb5b170e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 11:14:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b27adb629de943eda7bf2ffa9b7021df\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/76d49d2892cd4aebbaf7383f7c33ddfe : SUCCESS in 59m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdc356e8a9f04246a55ddac4fd47449f : SUCCESS in 1h 08m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbff714abc4a4b5bba8e0c72ad6e21ba : SUCCESS in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd96a6f76e524214bca6aba9b650ac46 : SUCCESS in 3m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4905acca69c14f01878392cc092a5a36 : SUCCESS in 5m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3584a6113fd0426097c746ae55e97ba8 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12f8f1438350456baa45192b2b91fe96 : SUCCESS in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a36e578f67e461cb525bd06b94b1823 : SUCCESS in 8m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c70843b160ba4a1daac7bac66a7c5c38 : SUCCESS in 8m 50s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e5dc86e9a650406fbeae2aed5c541e4f : SUCCESS in 6m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/74957efc316d4121b239617b856f610f : SUCCESS in 28m 11s","accounts_in_message":[],"_revision_number":1},{"id":"2adf1e538d13bb6cd8abd3c9daa98b0f3e3359a2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-03 14:21:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e183d5775a68fe0f19b239a510486c1cb2099d54","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-09-10 12:58:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bd418da4eafe4198ffcf48af9f8cf9bfe1def48b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 12:59:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6469d275a621e46c1ef4f744547a08fe4106aeb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 14:12:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d5bb8dc5f17446ab5cfae77e70d2806\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dc1faf4ccf2a4d5b9a5ee0f8760626ac : SUCCESS in 1h 05m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1ab5989960c4fe1bd1cb83f74842b80 : SUCCESS in 1h 07m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7f1704754964cf388e5354ba19f6a2f : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d365079939154a479ee095dc4dbcc8c1 : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a14941f106648c38480428837075397 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02355ff81e45412bb0af45610692f95b : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91dd463239f64c43976d60b5cc276589 : SUCCESS in 7m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/005d011ab27a46868ceaf0d60526415c : SUCCESS in 5m 58s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e56ce8fea635456d837359968ed7423c : SUCCESS in 5m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f928ee340e2742658904ce897520bb1f : SUCCESS in 34m 49s","accounts_in_message":[],"_revision_number":1},{"id":"513a9767bb3c6f1eba9b8017a4a9e8fa4a56fd98","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 14:12:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ad9dd3a37d902290761b2ea01ca8f5edfc386303","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 14:14:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5ca67d0ac8f48d99f0587d2ff1ed496\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbf6e7f7c88e4611b394fc30f3380095 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0b2be533bfe108e1ddde06b3952ed03afc9c781","revisions":{"b0b2be533bfe108e1ddde06b3952ed03afc9c781":{"kind":"REWORK","_number":1,"created":"2026-08-18 09:59:58.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/96/1001296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/1001296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/1001296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/1001296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/1001296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/1001296/1"}}},"commit":{"parents":[{"commit":"4f7a19df0cdad9cb50e92bc8a69be13543c4a3b5","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4f7a19df0cdad9cb50e92bc8a69be13543c4a3b5"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-08-18 09:50:28.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-08-18 09:50:28.000000000","tz":180},"subject":"Document config dependency in stage test","message":"Document config dependency in stage test\n\nThe test already relies on the default config `enabled-import-methods`\n\nChange-Id: I18b4b51a179b86c8e12c142cada16cc63bbc3539\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b0b2be533bfe108e1ddde06b3952ed03afc9c781"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b0b2be533bfe108e1ddde06b3952ed03afc9c781"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
