)]}'
{"id":"openstack%2Fglance~1001324","triplet_id":"openstack%2Fglance~master~I831506fd832104abc07c379e6b9cc0f173339bcf","project":"openstack/glance","branch":"master","topic":"refactor/glance-policy-remove-admin","attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2026-09-10 07:01:30.000000000","reason":"\u003cGERRIT_ACCOUNT_34510\u003e replied on the change","reason_account":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}},"34510":{"account":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"last_update":"2026-09-17 13:04:31.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-09-21 06:07:20.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"I831506fd832104abc07c379e6b9cc0f173339bcf","subject":"Remove admin role from service_api policy","status":"NEW","created":"2026-08-18 11:33:27.000000000","updated":"2026-09-22 18:29:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":12,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"9e281f4f3b9133955fc0e8c327d6566006b0c378","_number":1001324,"virtual_id_number":1001324,"owner":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-21 07:45:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}],"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-18 11:44:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-18 13:32:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 07:01:30.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2026-09-10 07:01:30.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2026-09-22 18:29:33.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"}],"messages":[{"id":"c6096ddb7223795ca7f020c15eed111848e75fef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-08-18 11:33:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21878b533e31c7e51c2847e5a9cb40d3ff715ade","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-08-18 11:33:55.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"451a2dff43ea71e4fd9ef571795089f875404fa4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 11:44:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c9569bd69a946a09e0cc928ef2efe87\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6d99565c3d374338b955808a553e5c7e : FAILURE in 5m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/08426f7d65ba4d79974cb3625493b2a9 : FAILURE in 5m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d844ecc38b8e4914ae8ccf0eb51bba2e : FAILURE in 6m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b4c3c52291deed51671d5ef52fa7c651bf60f37e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 13:32:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c13f9824b91d4bdabe9fcfbb37f369cf\n\n- grenade https://zuul.opendev.org/t/openstack/build/4f5136e3adf9483f9ccfa76346099955 : SUCCESS in 30m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/981874966e88407d983ada05c3e288d2 : SUCCESS in 30m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f463d58da5ec4c6d8fba1a9422cc3b93 : SUCCESS in 58m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a8672de7279489ca06d5181afafb30a : SUCCESS in 33m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1adc2079f784f47a92178feeecf9556 : SUCCESS in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f7820192b4194a248101422d73c3e31a : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c0e267e90d604749b70d1053716c96e3 : SUCCESS in 7m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ce945362895f41d588f435390404a5b8 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90005ebcb9284c59b9fd9defa59c6f20 : SUCCESS in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2da7e70f62d34f5dad81cd165474ec46 : SUCCESS in 3m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c730c3c00d224ca19cff6bf9e6ff3ac0 : SUCCESS in 6m 20s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/84decf286b8646eb9cc813cb87c8d34b : SUCCESS in 4m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d46ae541b78c46aab83a55a0fdbee050 : SUCCESS in 1h 51m 47s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/761f04fd752a4cd687df9a5e6cc356c0 : SUCCESS in 1h 33m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/82fd4d92148d405c9e712e2c1ab48c24 : SUCCESS in 58m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c2a6d4de8ef4ca98c0107d50767caa9 : SUCCESS in 26m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/97fe480cc9b64a8c96ae354cecfb1c88 : SUCCESS in 1h 28m 35s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/f4e3a8f6d678424799b50e7360193d5f : SUCCESS in 46m 54s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7411b41cec564ad19990d9c06ae817ae : SUCCESS in 35m 57s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/bf9809d21d404868b057257c31ba7a6c : SUCCESS in 49m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"325612d34970fb2f2e2d935e1b565f6224cf7676","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-09-09 14:16:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"ec18703bb9b371f86755a2b9404c785e3a8a6309","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 16:20:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c85aacea0fb04cb2b093b4a7b513628e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2d00aad1e0fa4ea9a618a664fc6794fe : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5ccc2da17cb8438c84b554a514d08679 : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/776d76ebf33241e6a997e77bb99b79c0 : SUCCESS in 13m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"de4b4af1557a4691d3e939125694819292d19f03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 16:49:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf5c95bd5224456db340ea51fb8a28b3\n\n- grenade https://zuul.opendev.org/t/openstack/build/9bbed7e2f73842e39b38b6934d47f26c : SUCCESS in 1h 11m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bf6e3985f25a4f80aa10d543d0c09391 : SUCCESS in 59m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c25d25ea11014b559c076818e3d78aed : SUCCESS in 1h 33m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24d60e5fae32411ea1c8ccac2bbd0c07 : SUCCESS in 1h 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a250e1a75dc40aaa2858b9598e70141 : SUCCESS in 6m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/281cdc38dd224dad8f1d161f5d50adcb : SUCCESS in 5m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/332ad5caa6084504baeb6e3c21817d05 : SUCCESS in 5m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1579435cc3f44cceb7fb9a1927d7405e : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f570262c482438787882f2a92253ce6 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bd8d683e74f49668dc3738965e4474b : SUCCESS in 8m 42s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/15e2b0e632ae4a30bbc794b9367706ee : SUCCESS in 8m 17s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4f6ffa99b1ce4e32aa05a705e0aa3554 : SUCCESS in 8m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cccf4ddb183e47e0a83ab41700f49fe0 : SUCCESS in 1h 48m 11s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d74f57b47bba44c69adad4e6efaa1684 : SUCCESS in 1h 33m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c0af671ac6434658a454b6b50786ae74 : SUCCESS in 1h 51m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/126c238e04704f2c9dbd07ac465d23b0 : SUCCESS in 1h 07m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ebb081550f424f9baac3f8274191b559 : SUCCESS in 1h 41m 16s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/176485248cb94320b33fd6961b33698b : FAILURE in 36m 58s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/de9f631a4bd14a5598fdbe2e047ea4b9 : FAILURE in 31m 06s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/af6bfa34bea44469a757e0e5baa2c793 : SUCCESS in 45m 01s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/739b50a460dd443da34f174b767301ec : SUCCESS in 53m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f7d9ca313ff4906d4254ff1f5d6870d8808385ae","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-09-10 06:58:11.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"3e5497e144ae452f8559ef15137152d443a5a02b","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-09-10 07:01:30.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"11f62c15664fec99ada8dff4fdfc33149e3680bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 07:15:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91434b4d49d846f98f1c25a982cae5fe\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b7f476a2a73549389ca328e8a1c79e05 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/75e8052131ef495987dcf79de37611ff : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c816b17a6f0e41e487b5a650b8f27331 : SUCCESS in 14m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f56f27b1bcaf69e923a37b15803af0428c2ed803","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-10 07:26:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"be453e893f7dba6973dcff5836161af97b90df68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 09:12:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/318599c88b7f4b3b810078c5356c35de\n\n- grenade https://zuul.opendev.org/t/openstack/build/75719ff7e66f4aad88fc7eccd65b2c25 : SUCCESS in 1h 07m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b7e69a58ecde416195ab84921a817c34 : SUCCESS in 1h 09m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/52f13910e93046b4afd02c430c16e5ef : SUCCESS in 57m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df32676052ba4cbb87f44f9eb2f39af0 : SUCCESS in 33m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72446d0701a1462fbe6af2e35fda7af9 : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8dfc95aba22f4f76ad38e464b4bdec51 : SUCCESS in 5m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/afbd85da5c254c6db2120d5f5aac5bd5 : SUCCESS in 3m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d0d205c299ce4536bd0b28c639ed1a22 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cdfc7c8c874454caa7b63dbb3213012 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8b3d7e475f243d6b00d3f428ca5e30c : SUCCESS in 8m 38s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/dbfe374989d442ff83618f49ec5aa560 : SUCCESS in 8m 35s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b4166a0ec2d1402cadb3cb3d555cce65 : SUCCESS in 10m 18s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c09e173ce8d14f68857887fe706a8f66 : SUCCESS in 1h 31m 59s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/62254dcfeed34b818a474ed54845b99c : SUCCESS in 1h 50m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b8756dffa62742ecbdca449777de1a62 : SUCCESS in 1h 58m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c6e1d7e5d9f48bfa54fc4524ed9ea3d : SUCCESS in 27m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f162fe0a2b844519b203a6039f207bfa : SUCCESS in 50m 34s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/d60876479877499d810c7c3e95e7d522 : FAILURE in 30m 19s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/0a71aecf3b3040878d14203e9de7843f : FAILURE in 30m 46s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/380d1bc65e46410087a37a1a2471e702 : SUCCESS in 37m 20s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/f92e358f58944868b9efa9d4e680224b : SUCCESS in 29m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9507d2a7d7ea15bc1a081b2daf653cb30294e893","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-09-10 09:17:34.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"efd0d912b1b87d74836232a0418c64b842aa87ed","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-09-10 09:46:10.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4d76e8f2b1cf1ada56370b34d126effbd0fb5e8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 10:15:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd75d9ac8cdb4a1ab025791036d3b4fb\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b213b9f4cf2d451ab63fb998a014ff22 : SUCCESS in 9m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3205ca8006774e49bf7f1ac43e4669e5 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5380c3fc3c5d41a69f3266db15395197 : SUCCESS in 14m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a13f7c9ec83bb65ff32671c541ae5a5e67a141a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 11:20:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/659cdaf1520943d78fa2cef8bd3912a9\n\n- grenade https://zuul.opendev.org/t/openstack/build/ba60f549d2f34fcd9bd92e821c222b88 : SUCCESS in 30m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b17e00bb15654811948702b5531c462b : SUCCESS in 53m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/be32268e98ca446db9b254172f78fc9e : SUCCESS in 52m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b2f541653c34125984d73cb325bae3e : SUCCESS in 44m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3effa4fe692439985223f8a0583584b : SUCCESS in 7m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d2ccc18702224a7a90e3d6404a59cb31 : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c45de7da684431eaedbdcf9664ae10d : SUCCESS in 4m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8c56fcef5e3c4d4d965692d43dcffae8 : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33233b168f094d759ad79d211df59a6c : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0e3d0e59e924170bef294f4cbc550fa : SUCCESS in 9m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5d9efa3215894d6b92a82283fa968a6a : SUCCESS in 7m 59s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a7d511a5633f4dbdb425c16d5e07fbe6 : SUCCESS in 7m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/006236aa2a31463ba933bd1dac6d015e : SUCCESS in 1h 19m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7d02ed67f34b43d2b06fc789a6619984 : SUCCESS in 1h 55m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cc320e64b20f4f5da09b27774d0e74f7 : SUCCESS in 1h 51m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79115c6ccc4c4f679502966c9690d732 : SUCCESS in 26m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0dddd1209aa04fbdbc7154eda5b3eec2 : SUCCESS in 1h 41m 53s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/fc5e776d19d84b6680bb68ffa1c66935 : FAILURE in 17m 07s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/25658f85605240488f40f70c790d1da3 : FAILURE in 14m 52s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e3fdb3a1c8d1499a895ed6bcd8d21ad4 : SUCCESS in 22m 52s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/677cb2718507421f8b1a0c15751aaf28 : SUCCESS in 25m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"df7aaaad55d00fd4d887bb1287e79419497818d4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-10 11:45:53.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4586c1bf3362b13c83f4729f1f35d8c99de0ec91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-09-14 14:59:10.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"e6d0e04488b8afbe7d331a363bf738bc29ae97ab","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-09-14 15:00:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a3f6e666e93f36996f0af0f1343e0e184e7694a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 15:09:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b936d94b053c4fe7a8facd783ab54ad6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b3cc1fe45c274efca32a50cd18d38b97 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/555c937932f3459eb76def067eddbe71 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e2a8c6c4fd8b424f9a7831929f564159 : SUCCESS in 8m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"82015902603065dfc2fbb116a88c04eb56c91b92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 17:18:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cd4760285234cf383e26e222164c0fa\n\n- grenade https://zuul.opendev.org/t/openstack/build/d9da3fe375b4499f8c2ecc08fe1d7150 : SUCCESS in 1h 06m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fc97d559089c41cea0f6d006a1ed4359 : SUCCESS in 1h 01m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ad92100cdb2d44299282e9ec5445d06c : SUCCESS in 1h 30m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/136134a9679246398975946ee3ba9596 : SUCCESS in 1h 02m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0883780ed27d451fa0a4296dd71806ed : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a316385652d646aeb07d85c20c63c2b3 : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4cd3f6126f9e42d49b9e6f1c4a60391c : SUCCESS in 4m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4b243b89a9ff4471bd776c5eb3255e98 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef2f13dd1ccb4f23a5d3701c058fd2ca : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bf55fa231524535a6566130e648b940 : SUCCESS in 9m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/11a13e3ca57c40b380fa988f1defcdf1 : SUCCESS in 5m 18s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b8e93f6998e64920a1e434e6d3101e3d : SUCCESS in 8m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/445a2d036e294749a2d16c77c559bac8 : SUCCESS in 1h 40m 01s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bd92b2e672c94589b8f9b49c02a8a858 : SUCCESS in 1h 57m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/289b624577684a0abbac128cf36590e7 : SUCCESS in 2h 17m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7413be812664bad87b83e9a40716f0a : SUCCESS in 31m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c2348dc435e44eafbb31d11ff03546e9 : SUCCESS in 1h 26m 31s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/82efd7be989e4db9b95950d9d46f022f : FAILURE in 29m 14s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/ced44211b29a43a0884bd7c68cf66799 : FAILURE in 27m 48s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5748d49bbda349cf899a889557ce90d7 : SUCCESS in 20m 20s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/22665d8c41f644d1b1bf5b372a71198f : SUCCESS in 56m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bcb97053ac0e1a3324607f0d946e2dbd5f54cd07","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-15 06:50:41.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f063cb23ae20f1b476c72a9d231583c875c2ac81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-09-17 10:53:15.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"d16af3363880e84234efd5be85c1a860bf5afb61","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-09-17 11:03:46.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"fb6f5b1675e723a8b7b9cc4df9b6d0c924c75a3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 11:39:03.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c2236db5b66428396777478971e7c28\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/93022d483dc7405c8fdae36ec57d22e3 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3a7c9bd43b5f454e8af54225175bfff8 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ba55b321ade64928973c180fbd3b081d : SUCCESS in 8m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ca9372e36bc469759f51fd3df01863257114a6e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 13:04:31.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/55814cbceb5a47dc975e0f9debd0c59a\n\n- grenade https://zuul.opendev.org/t/openstack/build/464283e1eecb4694afc62fa225112011 : SUCCESS in 1h 05m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/38a248613cd447078e372bcdcf904c0e : SUCCESS in 48m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e13867f665184ef0aa4ea5f307180211 : SUCCESS in 1h 06m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19788c45ccc84b7c89d7374e8607e819 : SUCCESS in 58m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12d429bc918b4cb582e25583bbcc77dc : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2524195db3224992bebf188204008dd9 : SUCCESS in 5m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/551bff5805ed4225866fc35c9455d55e : SUCCESS in 5m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1c065192399b48f78081f1b46edd0656 : SUCCESS in 10m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f631a0dfd3ae4c54872b20faa9db1d7c : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fda2660a62514897b0b184aff9d8e7da : SUCCESS in 9m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/493c85637b06498793bc195c85111817 : SUCCESS in 7m 46s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d4115ab695c54b5cb7f447c4d2b53cba : SUCCESS in 7m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/aeede591e08147f4b4a9b729b69ce0d3 : SUCCESS in 1h 36m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1f679b8402fb4c4f98e7f8c7175b471f : SUCCESS in 1h 48m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/36a05d53ea6f4076890304c86512d550 : FAILURE in 2h 01m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38f5110ce0f14b659d10ae53994ffced : SUCCESS in 1h 06m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b902d3f87d834f0fb25bc5fe440065fd : SUCCESS in 1h 27m 10s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/7cfb845ee7524e8fb16cca3f3f3f6647 : FAILURE in 20m 17s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/b31c8c5373d84482bd839ef8987c0a53 : FAILURE in 28m 44s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8edec26d63964a48b9f3a3b7c95a8677 : SUCCESS in 36m 02s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/98df30443c3349ee97f2fdcbe7e59a60 : SUCCESS in 51m 43s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9588db5ea208a16a71a4e5a1483f852362bcc650","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-21 06:07:20.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d5ace9e92b58b2b44e4c520baa4055c870afcefb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 06:21:37.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8b477213aed411a913c6bbc4e2fcdb5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c235e142af3b4c6197484c6eaccf8298 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/34427d31fbbd4ccfadeb1b3b495ba20c : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d427de7447ce40b2838f2b46cd7bf076 : SUCCESS in 12m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d4d5a4cb5b7926c993302ee425db283338284015","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 07:45:42.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc9757f5883047b2841a1cbbbb59312e\n\n- grenade https://zuul.opendev.org/t/openstack/build/155421ffc5d046caaf4827b44bd67b9b : SUCCESS in 1h 03m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5b140b5a258344d388ad101e47ffe172 : SUCCESS in 35m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d0df11d892564d56b927f3f9847f61df : SUCCESS in 1h 07m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d41f9229516b43e681186884a6219357 : SUCCESS in 57m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46cdc1cb8ac34e7cbcc4ea81bfeb5ad4 : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c957c92c85f41caafd2bedcbbe10b55 : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e0cf9e7534044a7acb287f0b5842b48 : SUCCESS in 4m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d81aa696aa04440bb74fd9378fab2dd3 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb4160347fd742f580507c3b67532816 : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c07b7a79d6ac46c496357d3b3e4d3b3d : SUCCESS in 8m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/92105cdb8fe043e4a7054a24604d95e6 : SUCCESS in 6m 21s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/36a12b78218447a4929920f8470d6e77 : SUCCESS in 9m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c49963f463b94952a192d7d99b66ec7b : SUCCESS in 1h 32m 57s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/99758d22dbab4df4be3a9134c6fe5f97 : SUCCESS in 1h 14m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c06ccce5f3774efd9208262e87ed00c8 : SUCCESS in 58m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60f21172d9b5467cb886f472321f648e : SUCCESS in 52m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c15631e5591e4df884d2d06f5228ba6d : SUCCESS in 45m 37s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/185b784af7c04f33b28d8faa8ab1e1eb : FAILURE in 18m 54s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/923536b72a9d40c18aecfb566bb22bae : FAILURE in 16m 36s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/41ea8011ce1244cda6f9bc3f9bb5190a : SUCCESS in 24m 39s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/67b84a43ba5d491b91f5e135d941b734 : SUCCESS in 29m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9e281f4f3b9133955fc0e8c327d6566006b0c378","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-09-22 18:29:33.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"209d3a3c3b5a68294aa26043fd9ee4643afdb4d0","revisions":{"505c5a7b0806200b5b19ffd53008ca3dff3c2a8a":{"kind":"REWORK","_number":1,"created":"2026-08-18 11:33:27.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/24/1001324/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/24/1001324/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/1001324/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/1001324/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/1001324/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/24/1001324/1"}}},"commit":{"parents":[{"commit":"720c21cf85554100171352879c4709cbe8d19517","subject":"Merge \"fix for 2 methods in metadefs policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/720c21cf85554100171352879c4709cbe8d19517"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-08-18 11:32:13.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-08-18 11:32:13.000000000","tz":180},"subject":"Remove admin role from Glance policy","message":"Remove admin role from Glance policy\n\nCinder uses the Glance service user to access Glance, so the admin\nrole is not required for this operation.\n\nChange-Id: I831506fd832104abc07c379e6b9cc0f173339bcf\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/505c5a7b0806200b5b19ffd53008ca3dff3c2a8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/505c5a7b0806200b5b19ffd53008ca3dff3c2a8a"}]},"branch":"refs/heads/master"},"633f7e925a4dc60303c466e2af687b2c4848e425":{"kind":"REWORK","_number":2,"created":"2026-09-09 14:16:31.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/24/1001324/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/24/1001324/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/1001324/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/1001324/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/1001324/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/24/1001324/2"}}},"commit":{"parents":[{"commit":"720c21cf85554100171352879c4709cbe8d19517","subject":"Merge \"fix for 2 methods in metadefs policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/720c21cf85554100171352879c4709cbe8d19517"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-08-18 11:32:13.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-09-09 14:00:37.000000000","tz":180},"subject":"Remove admin role from service_api policy","message":"Remove admin role from service_api policy\n\nCinder uses the Glance service user to access Glance, so the admin role is\nnot required for this operation.\n\nThe admin fallback was added in August 2025 (commit 301d8f580) as temporary\nbackward compatibility while Cinder added service user configuration for\nGlance. Cinder\u0027s service user support landed in commit 9dfb500d5 :\n   https://github.com/openstack/cinder/commit/\n     9dfb500d5bb4fb6523731d7053a0424280d11da2\n\nAffected policies:\n- fetch_image_location: service-only (no change)\n- add_image_location: admins must now be image owner\n- stores_info_detail: admin access remains via separate rule\n\nChange-Id: I831506fd832104abc07c379e6b9cc0f173339bcf\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/633f7e925a4dc60303c466e2af687b2c4848e425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/633f7e925a4dc60303c466e2af687b2c4848e425"}]},"branch":"refs/heads/master"},"ae06c156bac0859128fb26fd576ed45d69aae884":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-10 06:58:11.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/24/1001324/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/24/1001324/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/1001324/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/1001324/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/1001324/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/24/1001324/3"}}},"commit":{"parents":[{"commit":"720c21cf85554100171352879c4709cbe8d19517","subject":"Merge \"fix for 2 methods in metadefs policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/720c21cf85554100171352879c4709cbe8d19517"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-08-18 11:32:13.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-09-10 06:58:11.000000000","tz":0},"subject":"Remove admin role from service_api policy","message":"Remove admin role from service_api policy\n\nCinder uses the Glance service user to access Glance,\nso the admin role is not required for this operation.\n\nThe admin fallback was added in August 2025 (commit 301d8f580) as temporary backward compatibility while Cinder added service user configuration for Glance. Cinder\u0027s service user support landed\nin commit 9dfb500d5 :\n   https://github.com/openstack/cinder/commit/\n     9dfb500d5bb4fb6523731d7053a0424280d11da2\n\nAffected policies:\n- fetch_image_location: service-only (no change)\n- add_image_location: admins must now be image owner\n- stores_info_detail: admin access remains via separate rule\n\nChange-Id: I831506fd832104abc07c379e6b9cc0f173339bcf\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae06c156bac0859128fb26fd576ed45d69aae884"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae06c156bac0859128fb26fd576ed45d69aae884"}]},"branch":"refs/heads/master","description":"Edit commit message"},"11378ec387ab73bb91061e0ca1215c2966300412":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-09-10 09:17:34.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/24/1001324/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/24/1001324/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/1001324/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/1001324/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/1001324/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/24/1001324/4"}}},"commit":{"parents":[{"commit":"720c21cf85554100171352879c4709cbe8d19517","subject":"Merge \"fix for 2 methods in metadefs policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/720c21cf85554100171352879c4709cbe8d19517"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-08-18 11:32:13.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-09-10 09:16:06.000000000","tz":180},"subject":"Remove admin role from service_api policy","message":"Remove admin role from service_api policy\n\nCinder and Nova uses the Glance service user to access Glance,\nso the admin role is not required for this operation.\n\nNova has supported service users for Glance since (commit 6211009e5):\n   https://opendev.org/openstack/nova/commit/\n     6211009e557cf1c0addaf81e75e03939f8640a8a\n   https://opendev.org/openstack/nova/commit/\n     66bb1e77f2081ca37ac6211087afa9753ca90582\n\nThe admin fallback was added in August 2025 (commit 301d8f580) as\ntemporary backward compatibility while Cinder added service user\nconfiguration for Glance.Service user support landed in commit:\n   https://github.com/openstack/cinder/commit/\n     9dfb500d5bb4fb6523731d7053a0424280d11da2\n\nAffected policies:\n- fetch_image_location: service-only (no change)\n- add_image_location: admins must now be image owner\n- stores_info_detail: admin access remains via separate rule\n\nChange-Id: I831506fd832104abc07c379e6b9cc0f173339bcf\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/11378ec387ab73bb91061e0ca1215c2966300412"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/11378ec387ab73bb91061e0ca1215c2966300412"}]},"branch":"refs/heads/master"},"4cd20183d14bbf885905ee9bd1bbf22e7e49d16c":{"kind":"REWORK","_number":5,"created":"2026-09-14 14:59:10.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/24/1001324/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/24/1001324/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/1001324/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/1001324/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/1001324/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/24/1001324/5"}}},"commit":{"parents":[{"commit":"720c21cf85554100171352879c4709cbe8d19517","subject":"Merge \"fix for 2 methods in metadefs policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/720c21cf85554100171352879c4709cbe8d19517"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-08-18 11:32:13.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-09-14 14:58:44.000000000","tz":180},"subject":"Remove admin role from service_api policy","message":"Remove admin role from service_api policy\n\nCinder and Nova uses the Glance service user to access Glance,\nso the admin role is not required for this operation.\n\nNova has supported service users for Glance since (commit 6211009e5):\n   https://opendev.org/openstack/nova/commit/\n     6211009e557cf1c0addaf81e75e03939f8640a8a\n   https://opendev.org/openstack/nova/commit/\n     66bb1e77f2081ca37ac6211087afa9753ca90582\n\nThe admin fallback was added in August 2025 (commit 301d8f580) as\ntemporary backward compatibility while Cinder added service user\nconfiguration for Glance.Service user support landed in commit:\n   https://github.com/openstack/cinder/commit/\n     9dfb500d5bb4fb6523731d7053a0424280d11da2\n\nAffected policies:\n- fetch_image_location: service-only (no change)\n- add_image_location: admins must now be image owner\n- stores_info_detail: admin access remains via separate rule\n\nChange-Id: I831506fd832104abc07c379e6b9cc0f173339bcf\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4cd20183d14bbf885905ee9bd1bbf22e7e49d16c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4cd20183d14bbf885905ee9bd1bbf22e7e49d16c"}]},"branch":"refs/heads/master"},"209d3a3c3b5a68294aa26043fd9ee4643afdb4d0":{"kind":"REWORK","_number":6,"created":"2026-09-17 10:53:15.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/24/1001324/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/24/1001324/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/1001324/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/1001324/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/1001324/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/24/1001324/6"}}},"commit":{"parents":[{"commit":"720c21cf85554100171352879c4709cbe8d19517","subject":"Merge \"fix for 2 methods in metadefs policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/720c21cf85554100171352879c4709cbe8d19517"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-08-18 11:32:13.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-09-17 10:52:37.000000000","tz":180},"subject":"Remove admin role from service_api policy","message":"Remove admin role from service_api policy\n\nCinder and Nova uses the Glance service user to access Glance,\nso the admin role is not required for this operation.\n\nNova has supported service users for Glance since (commit 6211009e5):\n   https://opendev.org/openstack/nova/commit/\n     6211009e557cf1c0addaf81e75e03939f8640a8a\n   https://opendev.org/openstack/nova/commit/\n     66bb1e77f2081ca37ac6211087afa9753ca90582\n\nThe admin fallback was added in August 2025 (commit 301d8f580) as\ntemporary backward compatibility while Cinder added service user\nconfiguration for Glance.Service user support landed in commit:\n   https://github.com/openstack/cinder/commit/\n     9dfb500d5bb4fb6523731d7053a0424280d11da2\n\nAffected policies:\n- fetch_image_location: service-only (no change)\n- add_image_location: admins must now be image owner\n- stores_info_detail: admin access remains via separate rule\n\nChange-Id: I831506fd832104abc07c379e6b9cc0f173339bcf\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/209d3a3c3b5a68294aa26043fd9ee4643afdb4d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/209d3a3c3b5a68294aa26043fd9ee4643afdb4d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
