)]}'
{"id":"openstack%2Fglance~1002563","triplet_id":"openstack%2Fglance~master~Ie8928ea3d1e2c43468ef203ed6db1a96cec6ca71","project":"openstack/glance","branch":"master","topic":"cleanupreqs-requests","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-08-31 11:51:07.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-08-31 15:23:33.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ie8928ea3d1e2c43468ef203ed6db1a96cec6ca71","subject":"Requirements: requests is always required","status":"MERGED","created":"2026-08-26 22:44:12.000000000","updated":"2026-08-31 18:35:03.000000000","submitted":"2026-08-31 18:33:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002563-cleanupreqs-requests","meta_rev_id":"77160a3537a4b70b9c6a463cd24e97cbec09aef8","_number":1002563,"virtual_id_number":1002563,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-31 18:33:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-08-31 11:51:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2026-08-31 10:14:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-08-31 11:51:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-26 23:56:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-27 14:08:57.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2026-08-31 10:14:54.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"1d254de6decf6173a494eedbb523d202b6cf3d95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-08-26 22:44:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42bfd32e34b92ff7f7bd2bad5cb7aa5e78b8462e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-08-26 23:44:28.000000000","message":"Topic set to cleanupreqs-requests","accounts_in_message":[],"_revision_number":1},{"id":"01501ad9ee1b7c5bbf179cca094ad21b3b5475f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 23:56:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/712d89b040d54e038fe73dd04aa93f79\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/720d11c36a374701bd040bd672af27b5 : SUCCESS in 2m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f70b1301308f4f56851e7e6412494ac8 : SUCCESS in 2m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/30461358bdd0476687e3429e76fc9b90 : SUCCESS in 1h 06m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce335cd6a75440fbab1386a912131095 : SUCCESS in 42m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59901bcd9a4445d08148b19af289abcd : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4eb8e3ef15f4d81af36aec37bc8587a : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab0bd0f28b894ffbacd5dcc5726b80a8 : SUCCESS in 2m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c332c181336e47f29547d7e92695c4bb : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fdf635712e545e5803f0dfdfafd93d0 : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4326ebb8942543dfbdc9855c720bc5e7 : SUCCESS in 8m 19s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c5ec615ccc1b4b31b7f389d6a79e6a18 : SUCCESS in 4m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/028875937565447786651e6cb6036cbc : SUCCESS in 6m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9390c9df470c4282833cbb9f64b3ca1a : SUCCESS in 16m 19s","accounts_in_message":[],"_revision_number":1},{"id":"e6e793604f8b24b183eb9052c9f2bf8dc7bb00be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 01:27:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/470d25c653574664897ad8b6064008ce\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/14761f06999e4445a771ce1c828cc520 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/29d44ccde9884f02ab29552af50e919d : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5a6d55cc9c334908884af4e113828d77 : SUCCESS in 17m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84e06a4069edb7894536ba99550d513d9f842ede","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-08-27 14:08:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ad4c9d8ee7f78bce553e9a8f48785d6365d880ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 15:39:05.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/glance for 1002563,1","accounts_in_message":[],"_revision_number":1},{"id":"1a435dd00d594f00ced60f0e9f1a727b7913d849","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-08-31 10:06:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7bf7041be76a14f9dd2c966ea3778375da5c89d6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-31 10:14:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e39c10cff264848a86e80d922ae3d97e0530ace4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 10:30:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25dd76f7693d4a9aa87c67552230a895\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a46101b4ce1644d4b54a3cf1be8e5f62 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/62c8dc4a1919495e8496b87b3ef7c1c9 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0e1801a9c2204b75b30aa9d8588b23a8 : SUCCESS in 10m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8191c04be4390c30907ccd3e3a35a8c3d2d75748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 11:10:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b925223e9754504a99384a48b2061e2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cb7f4a2ec41146dc8d6fb75958696b3b : SUCCESS in 3m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1b7dec05e4a245ef85e567df3b66b17a : SUCCESS in 2m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a46e671013ad4615ab61808bcab82420 : SUCCESS in 58m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27dee01fec014e2d93e32fc7769ec98b : SUCCESS in 41m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4551f9130f76425e9eea0e2d5fcb421f : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4333051134294807967ad747ca41c850 : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f6aac786748b403194e3fe4ba89d796e : SUCCESS in 3m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8b10bfc98e454c40bd5c4043667101e0 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc5d457255a94be1bd84f8af98998662 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb6a9f87231b4768b6b4b3c073bd642a : SUCCESS in 7m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cdc2c4b98dcb434fbca329ebf19a3398 : SUCCESS in 6m 51s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/78022654813b45118fbae07a15f42bfb : SUCCESS in 7m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f08e1429aa4c40e594ee80e96a0ef315 : SUCCESS in 17m 15s","accounts_in_message":[],"_revision_number":2},{"id":"af555f4b59583d64e660cdf752236a60936bd523","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-08-31 11:51:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0a426d60ae68c8e351ef50751ca14661f1974252","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 11:51:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e8f64e479b4b923036ff2b054b665cfb3baea377","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 15:09:36.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/63a1397eb1c245d6baf9a990d1e957c0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e4613f448af944f59baffa6399321f6f : SUCCESS in 6m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bb48ae631c13411fa2d734d1a92805b1 : SUCCESS in 3m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/957fa22b35c446bc9a0fb01fc66cd09e : SUCCESS in 1h 06m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96f90cfeed944b47a4ee5941d791acbc : FAILURE in 23m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b60fb29f44c643028ddc9b171a39f68c : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/477dc2fc144b44d89152f179c622d8df : SUCCESS in 5m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a68016be41e4495784e93601a4676798 : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc282deda4434b3bb26427addce40532 : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdd3e06a620a48c8a9a6e3d64a36d447 : SUCCESS in 9m 19s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a625f93d8ed542849c0a474caca241a0 : SUCCESS in 7m 16s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fc74dd837fe047c488a8d6615698d1c4 : SUCCESS in 7m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/28297e82ab84409e92c8a09f9d78d992 : SUCCESS in 33m 11s","accounts_in_message":[],"_revision_number":2},{"id":"f92736feb952be396ba4bf79b3a6295478ee6ed7","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-08-31 15:23:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d4e7643de98e175bcc2b0f4aff1623410b4afd3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 16:09:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d74f72c8c51c4863a5bf31184e0d849f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c00d3968b71f48ada52ac98bbb3742b9 : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/51cdad15b3b24298bc253ec645d58b4e : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5cffe99bea8543b09b779e9d8b604b64 : SUCCESS in 15m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e50448257ec97c22576f37e61fe22f6c9bba6137","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 16:37:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a6b8d864f2c438b88b3c14349371011\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2e3fd8d62ad647f482603250bf559aed : SUCCESS in 3m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/80126f39d1c64626893aa42ad9904fb5 : SUCCESS in 2m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9dcc504229824e97a01d1625568c51e2 : SUCCESS in 1h 08m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1842e46d4024018a9729fd9b9e93756 : SUCCESS in 33m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7545c8b5717d4adca7a264d1c99c3693 : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd997f8e9e104eacbe20420c93f2a9dc : SUCCESS in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/276ea3f19c1a4715baab3347b7de73a8 : SUCCESS in 4m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b1da8bd1d68e42028f93e33116d61181 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3850aecff8b24ce5a8c0769125f257bd : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30a5fcef66b1417daf908cd2e2dede96 : SUCCESS in 6m 53s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/dca1a09b8cb64bc09b8388c4e9a644c7 : SUCCESS in 6m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0e653fd5746c4e059edc8035b8921321 : SUCCESS in 6m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5b764009cf74425baa3a2380c293f3ad : SUCCESS in 33m 59s","accounts_in_message":[],"_revision_number":2},{"id":"d9f91be4ac528b56bcb370d5d8b944332e50f34c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 16:37:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0eaeb4d6d63874202c033db1ee2ce01754c0b889","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 18:33:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14377154b3c7469d9156eaeef616418f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/35d9537c3d09409a8037027d484a7eab : SUCCESS in 7m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1af8e4d4034d46a787b4440a1c49f383 : SUCCESS in 3m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2699770c64674976a759ac137baedc1a : SUCCESS in 1h 07m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c071978e9514dcdada0e1c52ce8ce06 : SUCCESS in 55m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7516cc06a4cf42e590235b54dc3cbcaf : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c26c5891c51e4bdeb5f9395eb73b7e38 : SUCCESS in 6m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a579163d0444118bbe92c0f6b696b74 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce320ec571184aea8db7b43b232d4ec8 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cecc10b7afa74224a403a65664f9f0df : SUCCESS in 4m 59s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b114dd6586f1416f97ce09f4a716ae01 : SUCCESS in 5m 12s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e2b81926f6e6488881f49d43c10e2d8f : SUCCESS in 5m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0a7a382db5434358b9e49b5985d56a65 : SUCCESS in 36m 25s","accounts_in_message":[],"_revision_number":2},{"id":"b51dfc6f3f27db6b09efd2b32bae8ee7c7e043c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 18:33:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"77160a3537a4b70b9c6a463cd24e97cbec09aef8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 18:35:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74c9304a7b2c4eada2684b8d6bda9844\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/bea22553e7794a26b0be0e8ff58b9fe7 : SUCCESS in 50s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0dbae67e59f40038ae38308e9fb2976 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c2c9ec6e5834d00aab83c691fbd0f56 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4b369fc380fa17aa6a21b4a46530439bedee211e","revisions":{"3cc557be319682fda32b5637ad7240627a069702":{"kind":"REWORK","_number":1,"created":"2026-08-26 22:44:12.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/63/1002563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/63/1002563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/1002563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/1002563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/1002563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/63/1002563/1"}}},"commit":{"parents":[{"commit":"2699961fad7f17895abbcb82bb139843cebb4bac","subject":"tests: Ignore CHECK constraints in test_models_sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2699961fad7f17895abbcb82bb139843cebb4bac"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-08-26 22:42:29.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-08-26 22:43:57.000000000","tz":120},"subject":"Requirements: requests is always required","message":"Requirements: requests is always required\n\nThe requests library is used throughout the Glance code. It is required,\nand should be in requirements.txt rather than test-requirements.txt.\n\nChange-Id: Ie8928ea3d1e2c43468ef203ed6db1a96cec6ca71\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3cc557be319682fda32b5637ad7240627a069702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3cc557be319682fda32b5637ad7240627a069702"}]},"branch":"refs/heads/master"},"4b369fc380fa17aa6a21b4a46530439bedee211e":{"kind":"REWORK","_number":2,"created":"2026-08-31 10:06:59.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/63/1002563/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/63/1002563/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/1002563/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/1002563/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/1002563/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/63/1002563/2"}}},"commit":{"parents":[{"commit":"602b6d70307596290f77a8a533f8531d9cd4534a","subject":"Test requirements: remove testresources and testscenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/602b6d70307596290f77a8a533f8531d9cd4534a"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-08-26 22:42:29.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-08-31 10:06:44.000000000","tz":120},"subject":"Requirements: requests is always required","message":"Requirements: requests is always required\n\nThe requests library is used throughout the Glance code. It is required,\nand should be in requirements.txt rather than test-requirements.txt.\n\nChange-Id: Ie8928ea3d1e2c43468ef203ed6db1a96cec6ca71\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4b369fc380fa17aa6a21b4a46530439bedee211e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4b369fc380fa17aa6a21b4a46530439bedee211e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
