)]}'
{"id":"openstack%2Fglance~1002566","triplet_id":"openstack%2Fglance~master~Ic703fb7e18544f225dd300e62785cd4e97ddf372","project":"openstack/glance","branch":"master","topic":"cleanupreqs-testresources-testscenarios","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-08-27 15:39:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic703fb7e18544f225dd300e62785cd4e97ddf372","subject":"Test requirements: remove testresources and testscenarios","status":"MERGED","created":"2026-08-26 23:07:30.000000000","updated":"2026-08-27 15:41:49.000000000","submitted":"2026-08-27 15:39:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002566-cleanupreqs-testresources-testscenarios","meta_rev_id":"08c514ea1eac36c19778a8410eeed9299fa3bdca","_number":1002566,"virtual_id_number":1002566,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-27 15:39:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-08-27 14:08:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-08-27 14:08:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-26 23:40:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-27 14:08:38.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"327055f5248296da20a1299f93eab0c93ffd6747","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-08-26 23:07:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c49bdc0c0eb908bbcb0d7b25126fe840a50525d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 23:40:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a9462ae73354e448abb0437dc4216ba\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/35fd57775a88435293b3fd2f67c52601 : SUCCESS in 3m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/defedb5418944e37b93dee0bbe299b96 : SUCCESS in 2m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cf5ea2e066bd40348451251de33fa1dd : SUCCESS in 31m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/20673c1eb4fa4686b2d46997399fad5c : SUCCESS in 30m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a52ad250827340b881e9cd70b223916a : SUCCESS in 2m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/48153fa1ad434043a43af6321dd39519 : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/857e4d2e154f4802912f684ab6cd4474 : SUCCESS in 2m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/39daad7977544458b3d4c2239bb6daf6 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6139d8eec8d49c0b7d8f96ad166941c : SUCCESS in 3m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/148cba57ff4a4e06ae6ae27b35fe37f9 : SUCCESS in 7m 51s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d56f90aaf9d249ceb5ce8cdfd3613681 : SUCCESS in 6m 00s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f126be71bead44e9977871374291e8cd : SUCCESS in 7m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/721935a65149410aa525c9a64706a5e8 : SUCCESS in 17m 07s","accounts_in_message":[],"_revision_number":1},{"id":"dae495a0f535f8129ed8adfc4d14cc259e8af410","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-08-26 23:45:01.000000000","message":"Topic set to cleanupreqs-testresources-testscenarios","accounts_in_message":[],"_revision_number":1},{"id":"389505a22cbe753a92d83425e7b8315f7b456895","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 01:40:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4e13c3369c84d65ac5770215ff6108d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ea324c40b7f34107bf69df6391bc1277 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/24e2fb13168a49a5beba67e7c82e9f4d : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b45ef8e69f2a4b2c84cb32cadebbc339 : SUCCESS in 15m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c6692985a61a449002b4fa009774d6f5e9180717","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-08-27 14:08:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35fce459448f68937c3031e79de9c795fbd8d073","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 14:09:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd15df92b03c8bb5df353b12d152638225f42921","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 15:39:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e52d8e4753a4d36b7d3cecffa7c9716\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5c2724bd31b64cc8ad34bfbd9c533012 : SUCCESS in 4m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1c04f75c0bc146af8d475a818e896e9b : SUCCESS in 4m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/53a9fda04cf842259370e40a5ca7aa9b : SUCCESS in 1h 06m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68291574745a4fda9d5768b165db60f3 : SUCCESS in 27m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc16294bcf4b4e4e84b54d50fdd97538 : SUCCESS in 5m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ca1f78ffbf8d4e8096de3a44c4cd0828 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2595b1f6a7f14b498ebc73652e313358 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/035c1e1c2ebe4e4d9bbccd193ca1c1bb : SUCCESS in 7m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f65e1e354dc9460eb7c5d7b4382acfb9 : SUCCESS in 9m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cbe8af5e607d426980fcb74b01ce5da6 : SUCCESS in 5m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d2b1122b7e5645ca9d8772bb2cf3e28f : SUCCESS in 7m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0cabcc5408884421ae9a6fcb0c4f2d74 : SUCCESS in 30m 00s","accounts_in_message":[],"_revision_number":1},{"id":"40b59638da0017edbbb51da07b3adb116e8c3373","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 15:39:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"08c514ea1eac36c19778a8410eeed9299fa3bdca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 15:41:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0ea121e63ab4e8a91f6848faa29ba4f\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6e61eaab6c934b78ade1e135b1787052 : SUCCESS in 57s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1279f61ad4a4aafafd02d021c8abe97 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82e69a7d9e9c4f7d920fc64360616314 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"602b6d70307596290f77a8a533f8531d9cd4534a","revisions":{"602b6d70307596290f77a8a533f8531d9cd4534a":{"kind":"REWORK","_number":1,"created":"2026-08-26 23:07:30.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/66/1002566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/1002566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/1002566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/1002566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/1002566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/1002566/1"}}},"commit":{"parents":[{"commit":"2699961fad7f17895abbcb82bb139843cebb4bac","subject":"tests: Ignore CHECK constraints in test_models_sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2699961fad7f17895abbcb82bb139843cebb4bac"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-08-26 23:05:18.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-08-26 23:06:40.000000000","tz":120},"subject":"Test requirements: remove testresources and testscenarios","message":"Test requirements: remove testresources and testscenarios\n\nThese libraries were added to test-requirements.txt in\n693b50b2e79c9f3a00aa9311651972d9e67e1bcf because at the time, they were\nrequired to use the oslo.db fixtures. Since then, oslo.db has added them\nin its own requirements.txt file (see\nacd1ceb69906b8357ec7c60257651d4731aad08e in the oslo.db repository).\nThey are now safe to remove.\n\nChange-Id: Ic703fb7e18544f225dd300e62785cd4e97ddf372\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/602b6d70307596290f77a8a533f8531d9cd4534a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/602b6d70307596290f77a8a533f8531d9cd4534a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
