)]}'
{"id":"openstack%2Fglance~1003066","triplet_id":"openstack%2Fglance~master~I1d5a8bb79393add54d60fbb781910be42636e8ee","project":"openstack/glance","branch":"master","topic":"cleanupreqs-oslotest","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-09-22 09:23:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1d5a8bb79393add54d60fbb781910be42636e8ee","subject":"Test requirements: remove oslotest","status":"MERGED","created":"2026-08-31 15:00:43.000000000","updated":"2026-09-22 09:25:24.000000000","submitted":"2026-09-22 09:23:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003066-cleanupreqs-oslotest","meta_rev_id":"c6944d6703042dfab383d17952aebf067aa8e7a0","_number":1003066,"virtual_id_number":1003066,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-22 09:23:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-09-22 08:04:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-09 08:58:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-09-22 08:04:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-31 15:29:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-31 16:20:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-09 08:58:50.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"state":"REVIEWER"},{"updated":"2026-09-22 08:04:59.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"518cf5c203c0044851ed6b55ec923bb5fb1ab276","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-08-31 15:00:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62fe57dcc81bb123f671fb51954dbb9191ab8d16","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-08-31 15:01:36.000000000","message":"Topic set to cleanupreqs-oslotest","accounts_in_message":[],"_revision_number":1},{"id":"6451406a13937dde7bfcb6f831181488a3d651bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 15:29:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a888109c7a744128bcfa5d74bc814705\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/42045f334a284ed5a35ee7af0de6ba4a : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/863163f6249a4bc793d8d79d930a1ea8 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/223d2396ccbc436394870d3bea96ce0a : SUCCESS in 13m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db4b6cd1d91cafa2c62a9722979d590a749fd407","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 16:20:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/500523120079492bb2c6472853b40f8d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/adea4a618aa4462b8895858cffe826e8 : SUCCESS in 5m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a195985c38a94dc38425f71d134f00d3 : SUCCESS in 2m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c2be165535bb45aab4ee86e8d02df778 : SUCCESS in 1h 03m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33a7d67a8e30403b80b64e5df5dc73d8 : SUCCESS in 1h 04m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e57890f3e2a34060b8da62a29f466894 : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6879eb97bc9648d9af5456e919637391 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d26c14cf3a744daf883a727c169c847e : SUCCESS in 5m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4671693dc76446dca826ae53d91e0c2d : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b09a3adba9446698f0920f6ea384b8e : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b210d9cd0374ad39417a3912fa7e154 : SUCCESS in 5m 50s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cc08c03c43de4a4dbb6ac707c23a4420 : SUCCESS in 7m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5a51bdb4b5d04b20b325f79eaa451314 : SUCCESS in 4m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f35437ed3c2842c69ace1f144b90050b : SUCCESS in 19m 57s","accounts_in_message":[],"_revision_number":1},{"id":"3891f8f908f80f44a62d2ed48339876c70f51663","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-09-09 08:58:50.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"13d28db6351761037b73cd53e253167026edf7ec","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-22 08:04:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e7f7cc668c007cb81bb646bb0507f7abe6f1979a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 08:06:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6693418f0e9f0dd0ffc86488453e1919279e6b6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 09:23:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1b3b1e78f394ee2b9c25c0ca1c231d0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/282ade3465d64d71ae8ea100e81ceae6 : SUCCESS in 4m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8673fdd3a3804c33ac079b04cae10491 : SUCCESS in 3m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/35011874fac34701b1d8e335f19e7e65 : SUCCESS in 1h 10m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64c0254f875f4b67bcf8d57099abcbbf : SUCCESS in 1h 03m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2105a9a8aa0a4990b5fb6e09c1243f08 : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b7d40d1a564e4505b77b91125d79dbb7 : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c85c896b3c304ff9910f2c81e7973abf : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18eb7be616db4cc8923cd48ab6633ea6 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c45b6bfd1f4433ba7fd259a08f3e898 : SUCCESS in 8m 40s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c293c7278c5241119586c9f74378f683 : SUCCESS in 5m 57s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/823831b2e9f3458993ef31b218b9cb46 : SUCCESS in 5m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8ef86441b4164081a090cd0c2591b1fb : SUCCESS in 20m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c6c2125b41cf97680d8e2a6792ada6b0ec43c5f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 09:23:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c6944d6703042dfab383d17952aebf067aa8e7a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 09:25:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f428ac84e3874460a97b828cd1efb308\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/4e65ff5d1c9d4467b30c46c471fb5813 : SUCCESS in 1m 09s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4b83635807548a1ad72c0855675f7db : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57e2b9f3cfc54a0b83851567ed05c4ea : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b277d4590f0f2afae219d326736e8a94c3fcfaa0","revisions":{"b277d4590f0f2afae219d326736e8a94c3fcfaa0":{"kind":"REWORK","_number":1,"created":"2026-08-31 15:00:43.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/66/1003066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/1003066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/1003066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/1003066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/1003066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/1003066/1"}}},"commit":{"parents":[{"commit":"602b6d70307596290f77a8a533f8531d9cd4534a","subject":"Test requirements: remove testresources and testscenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/602b6d70307596290f77a8a533f8531d9cd4534a"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-08-27 15:45:48.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-08-31 15:00:33.000000000","tz":120},"subject":"Test requirements: remove oslotest","message":"Test requirements: remove oslotest\n\nThis has not been needed since f840dbefa591bcfef6344fa3eb91dfef6b8279c0.\n\nChange-Id: I1d5a8bb79393add54d60fbb781910be42636e8ee\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b277d4590f0f2afae219d326736e8a94c3fcfaa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b277d4590f0f2afae219d326736e8a94c3fcfaa0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
