)]}'
{"id":"openstack%2Fglance~1003069","triplet_id":"openstack%2Fglance~master~Iafd57e2421de6949db597349983b86161e7bd8eb","project":"openstack/glance","branch":"master","topic":"cleanupreqs-boto3","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-09-22 09:00:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iafd57e2421de6949db597349983b86161e7bd8eb","subject":"Test requirements: remove boto3","status":"MERGED","created":"2026-08-31 15:08:17.000000000","updated":"2026-09-22 09:03:44.000000000","submitted":"2026-09-22 09:00:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003069-cleanupreqs-boto3","meta_rev_id":"18e674d9b36debb97fad6ce3f9ba778f97f10628","_number":1003069,"virtual_id_number":1003069,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-22 09:00:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-09-22 08:04:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-09 08:53:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-09-22 08:04:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-31 15:37:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-31 16:34:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-09 08:53:02.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"state":"REVIEWER"},{"updated":"2026-09-22 08:04:36.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"c7f41252714d53f1e71bfa5566fa9bcd8228e62b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-08-31 15:08:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c1056ed67d64be0a2c51b14931928f83d4f67eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 15:37:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba729933375f43d1af950a176089cb06\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/45c73a770eb749a5a2562eb852d50ad3 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9e8b64298a3c434893567ca803f4e20c : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a15840245dff41a9bb98d8018ad3fd00 : SUCCESS in 15m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"629852b2e60edb2c186fd48c63c31380c64afd59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 16:34:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a3b7e3651e7419e878412bcc1661b03\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ca0ed5fbbf4d4461bf4825a6aa335c96 : SUCCESS in 3m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3e373e948f06402ca37ab0f4ef5373f4 : SUCCESS in 2m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bc566a40936e406780b1da8af440d13a : SUCCESS in 1h 19m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75d56f8446ee4a6fb21e4a1cb4d4f12b : SUCCESS in 57m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73d39bec5cf74ae2847c69687b3011bd : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/42947dd41a4242b49a9c8c6db0387a05 : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/934c9dbb4c6b4c2d80d49e32cefd004c : SUCCESS in 3m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/74f55099d98c45529777d040cf76ccdf : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4a8483c44c24888877951642b240473 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5cb660ef356424da452115a8197fccf : SUCCESS in 8m 12s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a680e8ffeb0240e88631e56508dac8d1 : SUCCESS in 6m 15s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/20eccf0cdaf2405fb3003f75925366b0 : SUCCESS in 6m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/905f6193595a4efdbfa18238bf971db5 : SUCCESS in 35m 30s","accounts_in_message":[],"_revision_number":1},{"id":"ef97e4c801472dc23157846becd217892e1312e3","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-09-09 08:53:02.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"019e9a2272e54874f7e320dcb999cb878fe454dd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-22 08:04:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"36048e109736ef2fd44384a06c300ad0ecb4c0ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 08:05:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58efefb4c4b7928d8b086e4d802d1dca1e610dd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 09:00:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e2a9fedcb644de8b9c78f0abdf6fb10\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/65143a8295cb4d8880f8532bc1640229 : SUCCESS in 4m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cf05a5c39f9b49018f8883a25c9e1e5d : SUCCESS in 3m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/86c6968ba9c649ed909b387d04d8f8ee : SUCCESS in 53m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99f2abd02a7a419587aec6806916a74e : SUCCESS in 37m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e0fe756dc15446ea63e2f74fddd0460 : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a8a8af6aeb74479faa2cd3df0ac11cd7 : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/313b242986d846d7b75dccb7614b2b6a : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f13bd4aa3ee74dfa92c7b248cb11ec82 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/246b013f70804abe96ee60900da9228e : SUCCESS in 6m 09s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c45b56d7f55a40b9af5a576054555bb5 : SUCCESS in 8m 25s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/28defdc86b204c42afc5426b4f8fdebe : SUCCESS in 9m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/78693ace20f34261b185a3ce6d9bfe45 : SUCCESS in 17m 14s","accounts_in_message":[],"_revision_number":1},{"id":"5650227540e01646ab7ba7749b166bc2013cacee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 09:00:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"18e674d9b36debb97fad6ce3f9ba778f97f10628","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 09:03:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06886c8c1f6a49dc8d2fffad8dabb60e\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/058a9bd6af554aaba7a7b7b11664a4bf : SUCCESS in 1m 06s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ef33cfb91fb46cb8ed50a4c02835fdb : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/774c401e23834c749ab5cb34bcd51765 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9f05c26a4466a2cf4f2d7cf1652c2cf9ccf2610","revisions":{"d9f05c26a4466a2cf4f2d7cf1652c2cf9ccf2610":{"kind":"REWORK","_number":1,"created":"2026-08-31 15:08:17.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/69/1003069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/69/1003069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/1003069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/1003069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/1003069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/69/1003069/1"}}},"commit":{"parents":[{"commit":"602b6d70307596290f77a8a533f8531d9cd4534a","subject":"Test requirements: remove testresources and testscenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/602b6d70307596290f77a8a533f8531d9cd4534a"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-08-31 15:06:05.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-08-31 15:06:34.000000000","tz":120},"subject":"Test requirements: remove boto3","message":"Test requirements: remove boto3\n\nThis was added in 673666cbf4f5ee97b28f157f3b1e7115b72c5fee but I do not\nunderstand why. It does not seem required. This is also a dependency\nthat belongs in glance_store rather than glance.\n\nChange-Id: Iafd57e2421de6949db597349983b86161e7bd8eb\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9f05c26a4466a2cf4f2d7cf1652c2cf9ccf2610"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9f05c26a4466a2cf4f2d7cf1652c2cf9ccf2610"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
