)]}'
{"id":"openstack%2Fglance~1003075","triplet_id":"openstack%2Fglance~master~I9e85e653a81f00b0cb4fb8dea1627d07d0b981e8","project":"openstack/glance","branch":"master","topic":"cleanupreqs-privsep","attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-09-09 08:59:39.000000000","reason":"\u003cGERRIT_ACCOUNT_34510\u003e replied on the change","reason_account":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9e85e653a81f00b0cb4fb8dea1627d07d0b981e8","subject":"Test requirements: remove privsep","status":"NEW","created":"2026-08-31 15:32:27.000000000","updated":"2026-09-09 08:59:39.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f81381b8361e65f1ec2fe88029a2fb5b4cb70aca","_number":1003075,"virtual_id_number":1003075,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-31 16:11:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-09 08:59:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-31 16:11:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-09 08:59:39.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"state":"REVIEWER"}],"messages":[{"id":"40f4f2d5f05b46e6afbe77a795cdbfdce9d3a0e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-08-31 15:32:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a146defff4291038d8058e73911d432ba25a3e8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 16:11:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a9483a064c9466b8c8bbc3863fc38af\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8c404f4bd92a41f28b627c26264b9b0b : SUCCESS in 3m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ccf9f8896dd84b4286ce9d4e788ba1b9 : SUCCESS in 2m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6151079e5dc64365ba8eda714839af51 : SUCCESS in 36m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/966111d465314b1097718a2071f2356d : SUCCESS in 32m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e96e520aacf74b72860e0278737af83f : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d389942ca5844f5d9af26ee864285bbf : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d73e85f271140dd86a7ad60b165450e : SUCCESS in 5m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0bab1df5a0fb4477b10ac7f76f961457 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d695427d49d74476a5042593b353e6db : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87d597ac50f240ef8f1ef9f2d8350a8d : SUCCESS in 4m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b9a5f011b83c4e1bbb07a0fac4b2bf25 : SUCCESS in 5m 02s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5c71bcef47294a5ca966586412db441e : SUCCESS in 5m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e52db67a09ae45c09601c393b0052be3 : SUCCESS in 21m 16s","accounts_in_message":[],"_revision_number":1},{"id":"dd702fd4cda733fe26aefccee57b6604749a0f49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 16:17:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70c5716d46ee41e19c883a33b701eafe\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2ea1b4055398408282658bff365310d6 : SUCCESS in 9m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d641001604d14ac9b71ea415ec3b3908 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/800a4f8fb4ff443fbce2fe6dea5b84f9 : SUCCESS in 15m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f81381b8361e65f1ec2fe88029a2fb5b4cb70aca","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-09-09 08:59:39.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30465c59dd7db98e822a10d57b3cb717e41868e2","revisions":{"30465c59dd7db98e822a10d57b3cb717e41868e2":{"kind":"REWORK","_number":1,"created":"2026-08-31 15:32:27.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/75/1003075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/1003075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/1003075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/1003075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/1003075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/1003075/1"}}},"commit":{"parents":[{"commit":"602b6d70307596290f77a8a533f8531d9cd4534a","subject":"Test requirements: remove testresources and testscenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/602b6d70307596290f77a8a533f8531d9cd4534a"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-08-31 15:23:52.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-08-31 15:23:52.000000000","tz":120},"subject":"Test requirements: remove privsep","message":"Test requirements: remove privsep\n\nThis does not seem required by Glance.\n\nChange-Id: I9e85e653a81f00b0cb4fb8dea1627d07d0b981e8\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/30465c59dd7db98e822a10d57b3cb717e41868e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/30465c59dd7db98e822a10d57b3cb717e41868e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
