)]}'
{"id":"openstack%2Fglance~1003083","triplet_id":"openstack%2Fglance~master~I1e7ee89b1e40b41cb0aa0767fe54d4a9d6a54132","project":"openstack/glance","branch":"master","attention_set":{"38744":{"account":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"last_update":"2026-09-02 14:42:57.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I1e7ee89b1e40b41cb0aa0767fe54d4a9d6a54132","subject":"Update metadata definition for Intel TDX support in nova","status":"NEW","created":"2026-08-31 16:41:15.000000000","updated":"2026-09-02 14:42:57.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0db13a93a0a5ac9c963d8669d06ce6bf808ce464","_number":1003083,"virtual_id_number":1003083,"owner":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-31 19:19:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2026-09-02 14:42:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-31 17:02:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-31 19:19:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-02 14:42:57.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"fe49b3ea869269096268326bba6c6d99b735fa3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-31 16:41:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"913e8b7434a7bc36de77caf85826d994eb33cc32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 17:02:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4043e7c058ae493c9168bbea207cc5a0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ec1428d3b6954d459c62c6a030e5fb8f : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a1b7324ceb524bb6b473a7d1af302b6b : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3130b667ed044785a078ec845bb4d293 : SUCCESS in 9m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a5e11ab102728ebe358d81619eac790921547b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 19:19:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc057c82192845a59eb415fda434c72e\n\n- grenade https://zuul.opendev.org/t/openstack/build/9267d84354d94d5bb83f528efb3c6686 : SUCCESS in 1h 08m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ea4585f53b21455c86b64fd8f0e8c7c9 : SUCCESS in 1h 09m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ad703df053534cd6a2bd8fdb3b8db822 : SUCCESS in 1h 30m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a92e8b220573472cb47be6e649d2097c : SUCCESS in 1h 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1f72e4af9cc40a78f5043ca129ef92b : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b0faa99402cd456c8d508431c6ac450b : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/332fbe5a13c5499c86095d969f31ff35 : SUCCESS in 5m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6096f75142fa48658d6310934c2386eb : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eef8fc2991df457c9c8c3614def0d239 : SUCCESS in 8m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14778ff598b04ceaa6fc7ef5174bf018 : SUCCESS in 9m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/89ff5217c1874b16883974623cdaae4e : SUCCESS in 6m 22s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/add2caffd63f44ef93aabdfd47475515 : SUCCESS in 7m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/208647689a424719a809828b43e608de : SUCCESS in 1h 22m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4b525243351f4e7d9a8d4071dabc26d1 : SUCCESS in 1h 08m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1ba917fd873c436da81c0a933e932c6b : SUCCESS in 1h 44m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e869c44fed9448fd9e63f90554839c5a : SUCCESS in 36m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/acb332ce89624103b9a3cd3c92a7e430 : SUCCESS in 1h 08m 40s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/cc1a99f41bca4344b6e557191d6a022b : FAILURE in 31m 39s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/e3c7670eeb724bc6818c329e4f7016d9 : FAILURE in 21m 30s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/30085ac8f59d4020b66d25654d700021 : SUCCESS in 31m 23s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/5b258441f8c94c34b98949247af70f53 : SUCCESS in 48m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0db13a93a0a5ac9c963d8669d06ce6bf808ce464","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-09-02 14:42:57.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc6224148af2797a47e1537753805e827dac6d46","revisions":{"fc6224148af2797a47e1537753805e827dac6d46":{"kind":"REWORK","_number":1,"created":"2026-08-31 16:41:15.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/83/1003083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/83/1003083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/1003083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/1003083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/1003083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/83/1003083/1"}}},"commit":{"parents":[{"commit":"602b6d70307596290f77a8a533f8531d9cd4534a","subject":"Test requirements: remove testresources and testscenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/602b6d70307596290f77a8a533f8531d9cd4534a"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-06 14:55:03.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-31 16:40:42.000000000","tz":120},"subject":"Update metadata definition for Intel TDX support in nova","message":"Update metadata definition for Intel TDX support in nova\n\nNova introduced launching instances with Intel TDX memory encryption,\nwhich can be enabled by the new \u0027intel-tdx\u0027 memory encryption model in\nimage properties.\n\nUpdate the enum to support this new value.\n\nDepends-on: https://review.opendev.org/c/openstack/nova/+/999249\nChange-Id: I1e7ee89b1e40b41cb0aa0767fe54d4a9d6a54132\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc6224148af2797a47e1537753805e827dac6d46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc6224148af2797a47e1537753805e827dac6d46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
