)]}'
{"id":"openstack%2Fglance~1003263","triplet_id":"openstack%2Fglance~master~I239b1c6d7dd409b8abb5ab63b73305ee444081b9","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-09-03 03:55:28.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"I239b1c6d7dd409b8abb5ab63b73305ee444081b9","subject":"Handle filesystem store timeouts with HTTP 503","status":"MERGED","created":"2026-09-01 12:19:20.000000000","updated":"2026-09-03 06:33:35.000000000","submitted":"2026-09-03 06:32:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003263","meta_rev_id":"bd2d50bf7a6b49851915d7f049a93c6bee965838","_number":1003263,"virtual_id_number":1003263,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-03 06:32:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2026-09-01 12:25:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2026-09-02 14:40:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2026-09-02 14:40:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-01 12:25:38.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-09-01 12:35:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-01 14:08:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-02 14:40:51.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"509e8dcdc86e5c85983cfc98999acc7fecc4c0bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-01 12:19:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea0e95795920d5231e4466c2eeb800c074e1919a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-01 12:25:38.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/c89e3b3db9134d8e8b4e67b2db13279a\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/9e269328d6c54a7195f8ecd59ef50491 : SUCCESS in 5m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84a4752a724076af5d11417eea83f2bb5a20c19d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 12:35:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b300a2391fef44b0b1243520ab98cb09\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/203b667809014198b6e7dab0d770ec4f : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/48b9f6381afe48b288d4b9f35afb1642 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fea28593506b476c8310d7cfeb65620a : SUCCESS in 10m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67f832673b6cd6fc9400359d07acd64ede087719","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 14:08:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a451e28117a748d19991016397f5b00a\n\n- grenade https://zuul.opendev.org/t/openstack/build/dc9d68ba431e4a86b6f7126721623141 : SUCCESS in 1h 04m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/90360311d70e46caa1b3501fbd9a6dee : SUCCESS in 56m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d57da6483bbb4b60bf00236bfbcdf34f : SUCCESS in 1h 37m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4f53df34a564f02a39388c6ea3e4a59 : SUCCESS in 1h 01m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/096ecde392ce4d52850e044c8d3415cb : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b19e55cb04a43a882da11670789a15e : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2fb2b6ccffb542b2a6f4500eb1448bd9 : SUCCESS in 6m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ebde1a4d67da4024b73685c602c286fd : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0f24c5824144fd6b7fbd07ca0e3bec2 : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6113fdc1d3b34ef8b82a3855a81693d6 : SUCCESS in 8m 37s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a4027a8e29bf4ef0bfaef9ebf5cf29fa : SUCCESS in 7m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6abbb7d5184b4cadab00c3359a2727aa : SUCCESS in 7m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6531cee46f474b64bef0391b7990a975 : SUCCESS in 1h 34m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3916914101ce48c281001c16f0fa548e : SUCCESS in 1h 42m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/76c58edfb7104772862a1c6f839e905e : FAILURE in 1h 43m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7148a5aef7494a11bde81b70b58db572 : SUCCESS in 1h 04m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/731dd9d739a9436bb31bfa4afdd35040 : SUCCESS in 53m 22s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/9d2c68edc161409295043d111f93ea74 : FAILURE in 29m 04s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/3c3d0a4bb1b34590a78ff990e9ec4a13 : FAILURE in 19m 55s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4dec14dad5b4499ca8be0f63c3a2db40 : SUCCESS in 34m 05s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/997b597ffa884ae0bb5a1422d5db4cad : SUCCESS in 44m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b717c5f3e880feba1550372723b30172d49f83e3","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-09-02 14:40:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b5e03b2f3028b23868e8ccb9687061ddcd1f9fc1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-02 14:42:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c831b3137f0706db8b45c6b88491ba9e378715c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 15:18:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79a51431172b4f5e84f6b389be980184\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5ee97c0ba02142d6a30ce0cafd053b66 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/52c4457ea8714341b68fd8f1721ae216 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/13f6b3f4e5ac429385429bbb8eea640a : SUCCESS in 17m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c34b0255de05c3e4e531a3d74f0ad1e58cb240ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 18:54:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57ead97880ce4cc4ad6a6a3e8058c8ee\n\n- grenade https://zuul.opendev.org/t/openstack/build/b17cf2e2c12740fd86f317868320f8db : SUCCESS in 53m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4e40046323094ddd9f7e14f5a491c0cb : SUCCESS in 39m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5ce96b00b110409f8adf5363389ec001 : SUCCESS in 1h 33m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1e717cf395d4b6fb711da083cf019a6 : SUCCESS in 34m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6293d67558084bc9be6dfcb1917d87f9 : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/255dc8a73c47446b8f169c940a75cfed : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b69038c0e954d65a4b673beff0b4a89 : SUCCESS in 4m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/48533c719f3048cfad5eb8b19dd27206 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d1088b9055f4256acc666122f20d3e6 : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5713ab7854ad40a68615238a08301565 : SUCCESS in 8m 58s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0ff0e3557bf24b369c906a003db3b3fb : SUCCESS in 7m 59s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6bac036e5759440c8a07a20ae6d04696 : SUCCESS in 6m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4ae078b1b9a44a7098b8be631c526ab1 : SUCCESS in 56m 16s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/47d4262e38ea40b0a7f61b6f82a624ad : SUCCESS in 58m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/70b00b00603e408cb4c6806db04ddcb7 : SUCCESS in 1h 37m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aac44743b89e43808249224726b8ff06 : SUCCESS in 55m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7d088edbc9364d62bedb6be52980ccbd : SUCCESS in 1h 29m 25s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/6dc942f58b2548b8b2413a3994515b86 : FAILURE in 15m 38s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/756e6a5c678642f29f8e1269151f96ec : FAILURE in 28m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1ea926340c7f44da99867a9585bf0797 : SUCCESS in 21m 43s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/3aceaa00afa0458a807c1359f53c4b7c : SUCCESS in 39m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0513527f5b6c8404b0b204db70e75c4538ae292","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 18:56:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c594476f2aa00aef9bb85166eb73cec49b5723f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 01:05:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff4c20cb7ea240a883fe0c8f256de2bd\n\n- grenade https://zuul.opendev.org/t/openstack/build/16e86781fb7c4eecb8c993ded9e7f048 : SUCCESS in 1h 11m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d76445bd918d4483920003254f450105 : SUCCESS in 58m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bc597483b09f4ebaa18dd7448eca3f17 : SUCCESS in 1h 18m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c80cb7ce9f44b1d906c8b5c5708a582 : SUCCESS in 48m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66d58c4474454f5ab509cf1800ebfba8 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d52cf0efd8245af91448eb4ebc7489a : SUCCESS in 6m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/559aaef818f7464f8fda8ae1a6e62946 : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39734d71e98d41a08eee6e6186b83185 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e6645a20d74470f98a452c923453933 : SUCCESS in 9m 12s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d1e3dd817aac4a23be374406fb228b08 : SUCCESS in 11m 02s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/29f924543e7945fb842746d1f3a45f96 : SUCCESS in 7m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/48a20984da9f45e5afe9c60ec94e4e77 : SUCCESS in 1h 47m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d03923750834969a6f832b60bd43693 : SUCCESS in 43m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/99e81baba41e472eb51509e4ab48e2da : SUCCESS in 1h 35m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cfdb3048d92b4853ad60cf297f5f596b : FAILURE in 35m 37s","accounts_in_message":[],"_revision_number":1},{"id":"cae31122772863293369547d3ef87e678d6be6f1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-03 03:55:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c5a470bf75efccbf794e73ca6b37417449a2b0b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 04:09:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/955683b1162f4784bdd5b4b96bf378f7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2cbc6cda069d411ba95c03cba50e6436 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/678d2cb1cb2748c08de546dce809c13e : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e5dc7077640a41dfa1411cd470bea94f : SUCCESS in 11m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f445d2a5058464a6303944ebc3a19eb28de071d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 04:55:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/febbb0bd284e416f9168dc5b1d1ba423\n\n- grenade https://zuul.opendev.org/t/openstack/build/5f9341de8fc44f4397282cad71ba2d52 : SUCCESS in 36m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f2e1fdf332b94af995f611642f229e79 : SUCCESS in 31m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/846b7b0013024473b2b604692318a0bd : SUCCESS in 58m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ccff9fc48e145ee95fd7d1560dc8405 : SUCCESS in 39m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/452c37a4fea247c08d1794926cc50373 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/74809bd07e75416b98f527619fa300a6 : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ff6f23f6f8547cf987f384a22cc3c1f : SUCCESS in 3m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/00f037d394034d3b97b6325a0f9ac1e9 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5f26d49e6284384a58e34a0f991c7ff : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c89d4f13c6a4f428a783f7ecdd5f16c : SUCCESS in 4m 36s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/68464c58e69e4b3c97d60730eb61bc12 : SUCCESS in 6m 25s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b080075a82ef4e5cb59fdb9b857fa3a6 : SUCCESS in 4m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a71c0585b9df436b98769db28ff538f3 : SUCCESS in 50m 27s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c28a60b19c284bbab747253323b9ff94 : SUCCESS in 57m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b315677e88ab4bd3bf9a6323d1f96ff4 : SUCCESS in 56m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0cc38377d9c40cfa9cd7e08b9f0525c : SUCCESS in 32m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f4ce1ca2ea864b4c86b1038a7be5608f : SUCCESS in 46m 58s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/1257c29482e14ad2a2ea694f0611fbb4 : FAILURE in 15m 23s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/cbab581e47af41a78d14293123e81801 : FAILURE in 12m 24s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/17cad647f3884d5fb81f22b3db56ddcf : SUCCESS in 20m 54s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/d349d784e2374c049f6490b8ba474a19 : SUCCESS in 28m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a6ba20018bd8d5a9045c72c04e5117d422039a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 04:56:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ecb2b9e270d5f687e733383893eb8e958f0eb6e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 06:32:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b20fa4eb5cc473eba6f84fc3c06bf5f\n\n- grenade https://zuul.opendev.org/t/openstack/build/89b85f8c19dd494c822031ec358af4a7 : SUCCESS in 31m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6a9357ac22e642a9b5b1d903eaef5684 : SUCCESS in 41m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3ff2e228da90439b92399f8d6d3cf136 : SUCCESS in 1h 30m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fa6603a77f24569abd842f2c7b4b1e9 : SUCCESS in 29m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34d5e5bfa8964e17ac97c205dcbec5b2 : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ed899592f5a44d7296a1c24d3e3cd99b : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f5ec329b4cf04ea5b19247b785dffcfb : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db93f8235a1d4ce2841d858f11ffb227 : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c3a0e511cb34db6a5a37627c849586a : SUCCESS in 6m 28s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6d49607f7994475c86f4482cdbd8aee9 : SUCCESS in 7m 35s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c00205cf43434071a87c531588d46178 : SUCCESS in 5m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1ca79c1823e74331abd4cbe7944d2572 : SUCCESS in 1h 14m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ae444cfb67e491f93518546bff52763 : SUCCESS in 39m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/673eb51ec65341289b14c14746b753a0 : SUCCESS in 1h 31m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3315e8f57c9345dc8109e99ecdff7353 : SUCCESS in 21m 29s","accounts_in_message":[],"_revision_number":1},{"id":"330a7768f1051084a84fd3bcb06ea9d20c95698a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 06:32:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bd2d50bf7a6b49851915d7f049a93c6bee965838","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 06:33:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d0d047d1247499da7a73a3e3d283f32\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9676829f816049babef8b000bcd0b931 : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eba34420752c427496d0399b4e436031 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f846a9a5751a5be4e3cd324c9553b411b83b0e22","revisions":{"f846a9a5751a5be4e3cd324c9553b411b83b0e22":{"kind":"REWORK","_number":1,"created":"2026-09-01 12:19:20.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/63/1003263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/63/1003263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/1003263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/1003263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/1003263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/63/1003263/1"}}},"commit":{"parents":[{"commit":"7435eb4e6c156c17d8633a379bb6a9eed085fd6c","subject":"Default oslo.service to the threading backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7435eb4e6c156c17d8633a379bb6a9eed085fd6c"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-09-01 11:52:38.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-09-01 12:04:10.000000000","tz":0},"subject":"Handle filesystem store timeouts with HTTP 503","message":"Handle filesystem store timeouts with HTTP 503\n\nMap glance_store TimeoutError to HTTP 503 Service Unavailable across\nimage upload, staging, create/update, and delete paths. Document the new\nfilesystem store timeout options and add a release note.\n\nImplements blueprint filesystem-timeout-protection\n\nAssisted-By: Cursor (claude-4.5-sonnet) for tests and docs\nChange-Id: I239b1c6d7dd409b8abb5ab63b73305ee444081b9\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f846a9a5751a5be4e3cd324c9553b411b83b0e22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f846a9a5751a5be4e3cd324c9553b411b83b0e22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
